)]}'
{"id":"openstack%2Fcyborg-specs~978064","triplet_id":"openstack%2Fcyborg-specs~master~Ie87f5bed688c8cd1d924d95a81a509cdcf16f491","project":"openstack/cyborg-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-03-02 18:48:22.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-02-26 14:40:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"Ie87f5bed688c8cd1d924d95a81a509cdcf16f491","subject":"Remove sphinxcontrib.seqdiag","status":"MERGED","created":"2026-02-26 12:01:11.000000000","updated":"2026-03-02 18:49:19.000000000","submitted":"2026-03-02 18:48:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"978064","meta_rev_id":"6f38a2afd363f71302c785d705cbbc5da62dc41d","_number":978064,"virtual_id_number":978064,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-02 18:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-02-26 14:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-03-02 18:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 12:01:18.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-02-26 12:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 14:14:58.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"6dc5619a64327dbd08c21ce97913737e94173dcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 12:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27a17e673180495af3674895b38aef625c603b97","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 12:01:18.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"576980330ea5823aa4ed4c2c49ce7bdfc9d47c82","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 12:09:09.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/c72550750128465da3534a6570dade84\n\n- teim-code-review https://zuul.teim.app/t/main/build/10aefc94e50245eb9d1d6059aaabba2c : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c8a33b25920d6de85df18f2426390c1b27316d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 12:13:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d364c8746c84f168b1fba0a75e87aab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d079a2d343fd4dd0baa5de042669f929 : FAILURE in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70fbb8e05b5842aaa695d97c7f81122a : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4de3c63ccbc4e14e78a6071c152c7d650ffd24de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 12:53:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7493bebd46f59ef5f505a354a683830b8279e302","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 12:53:57.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c299f5fe1682bf440f453e3990379a8b51d66205","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 13:01:50.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/8a649b84141d484cbe1ae67780644198\n\n- teim-code-review https://zuul.teim.app/t/main/build/2cb12d4617704c3eb00e8e65634f0477 : SUCCESS in 7m 44s\n\nWarning:\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: Cannot read blob db7f57679f723725b4c8b31aba649db4e2b0432f\u0027\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path commit message in the commit\u0027\n  Comments left for invalid file commit message\n  Comments left for invalid file doc/source/images/","accounts_in_message":[],"_revision_number":2},{"id":"f56e9f5663bab4c19bb49cf5bc95b54787225d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 13:05:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f31016cd9f3f442197bdd516711fb87b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d425ff4de60a412bb9c5aee21b5e4205 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0e99b293c4402882edbf29732cc342 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"16cdc7130f8f37d043191096fd0411f9153bd7eb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-26 14:14:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abc9142af5d5c8ab893b0b4a53e1ea90accd7c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 14:30:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"de8e609c4d34be534254781e299fed31c1496194","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 14:31:07.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"62977ac9805b1ee4c7243a133f586a03a052a60c","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 14:31:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74ce742020b9c36190ff8436e622923d63f20f5c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 14:39:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/57b002c176b74437805f765aff516608\n\n- teim-code-review https://zuul.teim.app/t/main/build/02c9daf335aa484a9e612019819333c6 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7ddbd5fcec24ee2e30272e91949aa5bc1f0f2556","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-26 14:40:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ac0d03a926a2e0b6c09a7036cd27bfe3721274f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 14:41:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b068adf79da47499c22a86237d10ca1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/272d6f55c1e74b2cac1475d997721b06 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af9a3e935f742928fde936686366498 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"609f0803b6146c198a817ae735a87c9baf6fa60b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-02 18:39:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1220f03ed2078cef60a38d262fe41353eb613ea9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:41:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d7ecbd49f7453f82cb15243276cc9258f38954cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:48:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db820fef4f1472a959386d14fa6524e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ab3d627d8744be84114b559da50a08 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe402c6110a4ec3a7a6cb36d4c0d3fc : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"45572c1127728d9179d3886755f281c01b39d905","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:48:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6f38a2afd363f71302c785d705cbbc5da62dc41d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:49:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df93cfa0bcf2481b8874f94a88f9494f\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/84a9c6b4ca5b46188f09f356529cd222 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4a7d7833823105744f3134a16c850f9653ba68c","revisions":{"9433e61f01f432f049afc13479c978604790f7c8":{"kind":"REWORK","_number":1,"created":"2026-02-26 12:01:11.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/978064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/64/978064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/1"}}},"commit":{"parents":[{"commit":"08886300a2cf6c2384c764caa7dd28490f071821","subject":"Maintain and update approved bp disable-enable-device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/08886300a2cf6c2384c764caa7dd28490f071821"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 11:59:59.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 12:00:59.000000000","tz":60},"subject":"Remove sphinxcontrib.seqdiag","message":"Remove sphinxcontrib.seqdiag\n\nThe tool but it seems it is no longer maintained, with no\ncommits since late 2021 [1]. Remove it, capturing static copies of the\nSVGs it was generating so we don\u0027t lose information. The original\n\"source\" is retained.\n\n[1] https://github.com/blockdiag/blockdiag\n\nChange-Id: Ie87f5bed688c8cd1d924d95a81a509cdcf16f491\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9433e61f01f432f049afc13479c978604790f7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9433e61f01f432f049afc13479c978604790f7c8"}]},"branch":"refs/heads/master"},"9edd8e7c861b5f9e59e4f0e38a99d6374fb7d5b7":{"kind":"REWORK","_number":2,"created":"2026-02-26 12:53:50.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/978064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/64/978064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/2"}}},"commit":{"parents":[{"commit":"08886300a2cf6c2384c764caa7dd28490f071821","subject":"Maintain and update approved bp disable-enable-device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/08886300a2cf6c2384c764caa7dd28490f071821"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 11:59:59.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 12:52:11.000000000","tz":60},"subject":"Remove sphinxcontrib.seqdiag","message":"Remove sphinxcontrib.seqdiag\n\nThe tool but it seems it is no longer maintained, with no\ncommits since late 2021 [1]. Remove it, capturing static copies of the\nSVGs it was generating so we don\u0027t lose information. The original\n\"source\" is retained.\n\nFix warnings about footnotes not being referenced by removing the\nfootnote marker and simply leaving the link in the references section.\n\n[1] https://github.com/blockdiag/blockdiag\n\nChange-Id: Ie87f5bed688c8cd1d924d95a81a509cdcf16f491\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9edd8e7c861b5f9e59e4f0e38a99d6374fb7d5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9edd8e7c861b5f9e59e4f0e38a99d6374fb7d5b7"}]},"branch":"refs/heads/master"},"c4a7d7833823105744f3134a16c850f9653ba68c":{"kind":"REWORK","_number":3,"created":"2026-02-26 14:30:50.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/978064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/64/978064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/64/978064/3"}}},"commit":{"parents":[{"commit":"08886300a2cf6c2384c764caa7dd28490f071821","subject":"Maintain and update approved bp disable-enable-device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/08886300a2cf6c2384c764caa7dd28490f071821"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 11:59:59.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 14:29:01.000000000","tz":60},"subject":"Remove sphinxcontrib.seqdiag","message":"Remove sphinxcontrib.seqdiag\n\nThe tool but it seems it is no longer maintained, with no\ncommits since late 2021 [1]. Remove it, capturing static copies of the\nSVGs it was generating so we don\u0027t lose information..\n\nFix warnings about footnotes not being referenced by removing the\nfootnote marker and simply leaving the link in the references section.\n\n[1] https://github.com/blockdiag/blockdiag\n\nChange-Id: Ie87f5bed688c8cd1d924d95a81a509cdcf16f491\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/c4a7d7833823105744f3134a16c850f9653ba68c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/c4a7d7833823105744f3134a16c850f9653ba68c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
