)]}'
{"id":"openstack%2Fcyborg-tempest-plugin~1005315","triplet_id":"openstack%2Fcyborg-tempest-plugin~master~I6a1f907b90e46e566b9f92f48e00cccb5481b0da","project":"openstack/cyborg-tempest-plugin","branch":"master","topic":"2026.2-job","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-22 16:33:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-22 17:33:18.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-22 16:33:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I6a1f907b90e46e566b9f92f48e00cccb5481b0da","subject":"Add stable/2026.2 job","status":"NEW","created":"2026-09-11 17:16:23.000000000","updated":"2026-09-22 17:33:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"61c4dad8ad849a051bf8c13ca37b1fddd054e76a","_number":1005315,"virtual_id_number":1005315,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-21 17:21:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 17:17:18.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-09-11 18:14:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-09-11 18:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 16:19:58.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"df3fd4bc4318b144607f2db20f3e1bb99cf57ed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:16:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"989f1230ff007150fb8652296f5407fafd48fb15","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-11 17:17:18.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec34967c2ef582ffe65a8cb4bed58adeb3211e25","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-11 17:24:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/dfae1b4613d2465886a9e76db9ccd7c9\n\n- teim-code-review https://zuul.teim.app/t/main/build/490c988c1f8f433897627cdab5886391 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0f736ca1619400f07749a91685bd8214f14d89d9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-11 18:14:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b816a3c54a449fe5baca7c74ec9cefb41f0b71e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:17:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f76120de78a44b6999241602b38c6a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b283bd29990411cbb8aadd2b77ba42a : SUCCESS in 1m 53s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/920bb339bc444cbaba73968484d86fb4 : SUCCESS in 53m 22s\n- cyborg-tempest-2026-2 https://zuul.opendev.org/t/openstack/build/6379e84b37c84482860984e64f921174 : SUCCESS in 32m 51s\n- cyborg-tempest-2026-1 https://zuul.opendev.org/t/openstack/build/69d6f003f9f048e9853057ecbc3f1f12 : SUCCESS in 16m 29s\n- cyborg-tempest-2025-2 https://zuul.opendev.org/t/openstack/build/8bdb400f3591455ea9d457ed246be257 : SUCCESS in 33m 17s\n- cyborg-tempest-2025-1 https://zuul.opendev.org/t/openstack/build/531aa71943bb4a519f4a13c2fc3d4059 : FAILURE in 47m 01s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7c6c599ebf364519b3199db5af555d68 : SUCCESS in 47m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7684d99f0b14586cd69d8e8af0fe57a68857255a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-12 08:58:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f382d86684ff4637bc557c70aa1316fbf167aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-21 16:16:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"866c97eab14c04725442042819812ec8403f9790","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-21 16:16:14.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc00ac357cf673e8e0f0d3044d8cfba11994b504","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-21 16:19:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2a64adef10bc9c65c4be998ff32fcad4b417b90","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-21 16:21:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/17a74d7716e14b22aa2508b770e05110\n\n- teim-code-review https://zuul.teim.app/t/main/build/e0025bfccb7e4fdc9db0a8d20c6ebb1e : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2d65d600b23cc71c87f504309e23b3f19bcc959f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-21 16:25:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4603d3a0369c08eda8c76d0791cacdf61a2a91a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:21:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc1e03761444e798efa2f96703bc054\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c084ec4aa84506ac31f56f5420f4ad : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fa12d0c560f495ba68f6ad8d7a2aaa9 : SUCCESS in 3m 06s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a452e7ab550547b983e10b3283844cc3 : SUCCESS in 54m 45s\n- cyborg-tempest-2026-2 https://zuul.opendev.org/t/openstack/build/5672723124a648b18cbbb4b1fb0cd12b : SUCCESS in 31m 57s\n- cyborg-tempest-2026-1 https://zuul.opendev.org/t/openstack/build/93a742419a034326a717ed65ce71133f : SUCCESS in 30m 22s\n- cyborg-tempest-2025-2 https://zuul.opendev.org/t/openstack/build/a9e0e61a335f41ebb7367c0c168bab01 : SUCCESS in 37m 28s\n- cyborg-tempest-2025-1 https://zuul.opendev.org/t/openstack/build/f1c922a2be664a24887bb27d180fd335 : SUCCESS in 32m 58s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8ca93717c40b4272abeee0e78d0a0a07 : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e89925c3d66f439b6331d5a52aa646a19667988a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-22 16:33:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"61c4dad8ad849a051bf8c13ca37b1fddd054e76a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-22 17:33:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a86b496de0786083dd2f9eccbc69475b35d1a111","revisions":{"41093ce3b0624f7234553f57a3bc8a4952676782":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:16:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/1005315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-tempest-plugin","ref":"refs/changes/15/1005315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/1"}}},"commit":{"parents":[{"commit":"ae4fd72abc28b8cf032fac9279c502f0bf432380","subject":"Refactor PCI lifecycle tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/ae4fd72abc28b8cf032fac9279c502f0bf432380"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:16:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:16:03.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: I6a1f907b90e46e566b9f92f48e00cccb5481b0da\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/41093ce3b0624f7234553f57a3bc8a4952676782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/41093ce3b0624f7234553f57a3bc8a4952676782"}]},"branch":"refs/heads/master"},"a86b496de0786083dd2f9eccbc69475b35d1a111":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-21 16:16:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/1005315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-tempest-plugin","ref":"refs/changes/15/1005315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/15/1005315/2"}}},"commit":{"parents":[{"commit":"61ec6240f1e038f12809d14d9568d0000fc7eae9","subject":"Run build release jobs in cyborg tempest plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/61ec6240f1e038f12809d14d9568d0000fc7eae9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:16:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-21 16:16:10.000000000","tz":0},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: I6a1f907b90e46e566b9f92f48e00cccb5481b0da\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/a86b496de0786083dd2f9eccbc69475b35d1a111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/a86b496de0786083dd2f9eccbc69475b35d1a111"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ae4fd72abc28b8cf032fac9279c502f0bf432380","ours":"41093ce3b0624f7234553f57a3bc8a4952676782","theirs":"61ec6240f1e038f12809d14d9568d0000fc7eae9","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
