)]}'
{"id":"openstack%2Fcyborg-tempest-plugin~1006244","triplet_id":"openstack%2Fcyborg-tempest-plugin~master~I24c8dfa8a35b6e73af9d2ece9901afe19a8c0a66","project":"openstack/cyborg-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-09-18 17:15:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24c8dfa8a35b6e73af9d2ece9901afe19a8c0a66","subject":"Run build release jobs in cyborg tempest plugin","status":"MERGED","created":"2026-09-18 15:23:02.000000000","updated":"2026-09-18 17:36:11.000000000","submitted":"2026-09-18 17:15:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006244","meta_rev_id":"5b98715b8437fd9f9f1ac3e85befe49d6ed50038","_number":1006244,"virtual_id_number":1006244,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 17:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-09-18 15:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-09-18 15:47:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 15:23:06.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-09-18 15:47:10.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-09-18 16:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff7b1821a97115aea29229c0412b501369079ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-18 15:23:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c68dd777b566970aa3d9a7ba3782eeea48c0d98","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-18 15:23:06.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56f6ebe7f82df4c94942b2afda084e6400f46b3f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-18 15:47:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af725f7ab3dfc99af076fa9d07c5a5ba34928205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:18:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0b6af714f74537b77f53228f2d3d54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13b5d65d2ef47b3944414dcd8c8bd24 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ed736b4b6c841cfb780fa463b6810bc : SUCCESS in 2m 53s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/693dc67d39fa4605bfcc21702ae7ae3c : SUCCESS in 36m 17s\n- cyborg-tempest-2026-1 https://zuul.opendev.org/t/openstack/build/410dda48b4d84bd5bb1b6096133830ec : SUCCESS in 24m 17s\n- cyborg-tempest-2025-2 https://zuul.opendev.org/t/openstack/build/b3465f8f8a1746a1afbab5e0b68430cc : SUCCESS in 31m 13s\n- cyborg-tempest-2025-1 https://zuul.opendev.org/t/openstack/build/9fbc45e1161f4409821028e4c1850bd5 : SUCCESS in 33m 48s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/b860714d0be2473eb92a170b72a09324 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":1},{"id":"059b569f639d6d3bf32b86086d9ca90ffe99f4ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:19:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b24a9e0eb9810e162125e0d49c6cb3d1649f0e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:15:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9977bad4ec0a433aab6a736c4166b617\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58dfc183ac9640b691e1a1f6624bfd79 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a93f63e512c24cedba1ad6309b041e5e : SUCCESS in 3m 15s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a514250da6fa405f902d2a7f43e11d4a : SUCCESS in 51m 07s\n- cyborg-tempest-2026-1 https://zuul.opendev.org/t/openstack/build/51263f2388f841dd95c54ebc6b59de4d : SUCCESS in 31m 17s\n- cyborg-tempest-2025-2 https://zuul.opendev.org/t/openstack/build/b5f7dc0a37fd477b8317e36a92e2091f : SUCCESS in 32m 04s\n- cyborg-tempest-2025-1 https://zuul.opendev.org/t/openstack/build/abf5afab3bf44a9780ec55edccf5ed8b : SUCCESS in 32m 40s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/904b96a432ef44569395b76411912f70 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":1},{"id":"aa7b301ee3e0d137548b2b06408e98f0ebb6b361","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:15:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b98715b8437fd9f9f1ac3e85befe49d6ed50038","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-18 17:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/3660ce93bc364242b097a23657cdae15\n\n- teim-code-review https://zuul.teim.app/t/main/build/2015034bf61e4993b9e05d6d6b751a75 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61ec6240f1e038f12809d14d9568d0000fc7eae9","revisions":{"61ec6240f1e038f12809d14d9568d0000fc7eae9":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:23:02.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/44/1006244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-tempest-plugin","ref":"refs/changes/44/1006244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/44/1006244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/44/1006244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/44/1006244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-tempest-plugin refs/changes/44/1006244/1"}}},"commit":{"parents":[{"commit":"ae4fd72abc28b8cf032fac9279c502f0bf432380","subject":"Refactor PCI lifecycle tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/ae4fd72abc28b8cf032fac9279c502f0bf432380"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-09-18 15:22:57.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-09-18 15:22:57.000000000","tz":120},"subject":"Run build release jobs in cyborg tempest plugin","message":"Run build release jobs in cyborg tempest plugin\n\nChange-Id: I24c8dfa8a35b6e73af9d2ece9901afe19a8c0a66\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/61ec6240f1e038f12809d14d9568d0000fc7eae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-tempest-plugin/commit/61ec6240f1e038f12809d14d9568d0000fc7eae9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
