)]}'
{"id":"openstack%2Fcyborg~626057","triplet_id":"openstack%2Fcyborg~feature%2Fcyborg-nova-pilot~Icb92206b4f6bcebc4c0be219d3937a1f211cac82","project":"openstack/cyborg","branch":"feature/cyborg-nova-pilot","topic":"cyborg-nova-pilot","hashtags":[],"change_id":"Icb92206b4f6bcebc4c0be219d3937a1f211cac82","subject":"Changes for v2 API and Nova integration for VM creation.","status":"ABANDONED","created":"2018-12-19 00:24:58.000000000","updated":"2022-01-29 08:00:18.000000000","total_comment_count":96,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23b6b15489764cf04a221dfbcda4a4a27177efc3","_number":626057,"virtual_id_number":626057,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-29 19:33:26.000000000","_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":-1,"date":"2019-07-12 06:26:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},{"value":-1,"date":"2019-07-29 23:50:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-05 08:14:52.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-02-07 19:30:44.000000000","updated_by":{"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},"reviewer":{"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},"state":"REVIEWER"},{"updated":"2019-04-12 07:34:37.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-05-14 10:44:52.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-05-15 02:34:27.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-05-20 11:35:31.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2019-05-28 11:38:03.000000000","updated_by":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"reviewer":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"state":"REVIEWER"},{"updated":"2019-07-12 06:26:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-29 19:33:26.000000000","updated_by":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"reviewer":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"state":"REVIEWER"}],"messages":[{"id":"bc1dd033fb8e160ef0933778496d99bc884ed694","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-12-19 00:24:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0db60145ec12060d6b528043c2f0777db3d996fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-19 01:03:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/1/check/openstack-tox-pep8/950599c/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/1/check/openstack-tox-py27/3a3c0ad/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/57/626057/1/check/openstack-tox-docs/e897f20/html/ : SUCCESS in 4m 57s\n- cyborg-devstack http://logs.openstack.org/57/626057/1/check/cyborg-devstack/1deacfe/ : SUCCESS in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9845dccb62affbf1c2ca7646ee50d2484ad5df7e","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-12-20 01:29:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"208939ea2ccca590e27313fd6606b7e44ab6ce69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 02:27:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/2/check/openstack-tox-pep8/d28eb71/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/2/check/openstack-tox-py27/73e5d41/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/57/626057/2/check/openstack-tox-docs/1d6127a/html/ : SUCCESS in 4m 34s\n- cyborg-devstack http://logs.openstack.org/57/626057/2/check/cyborg-devstack/6db75bd/ : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed34de0826dae0e21c40faf1833e7701b411d208","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2018-12-31 03:19:13.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, let another core to give a +2","accounts_in_message":[],"_revision_number":2},{"id":"0c80b293a33c8e7efb8ceb59c28d563a5319d258","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-01-05 08:14:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7fff3c5317a9d0ce93a4ba5c6ea999a83a6d72e2","author":{"_account_id":11878,"name":"Rushil Chugh","email":"rushil.chugh@gmail.com","username":"rushil"},"date":"2019-02-07 19:30:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5ca1684d63c61f25cd2f31bf68b03c06e454a54a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-03-21 04:41:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fa20d04cffd9f5070701d3ac49253da3b3bf2f52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 04:53:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/3/check/openstack-tox-pep8/c46dfb3/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/3/check/openstack-tox-py27/f8654ec/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/57/626057/3/check/openstack-tox-docs/3b5938d/html/ : SUCCESS in 4m 38s\n- cyborg-devstack http://logs.openstack.org/57/626057/3/check/cyborg-devstack/dde3b02/ : FAILURE in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f62b45b1e6e282e5854cbd1b3989eca3890cebb5","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-04-03 20:47:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e65d26b31badc0fea33abd5f891df783e09290cc","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-03 22:36:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ce04a1bef60249510f24a873bd3196a522a7a585","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 22:57:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/4/check/openstack-tox-pep8/cef44a8/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/4/check/openstack-tox-py27/a4ab6b1/ : SUCCESS in 7m 46s\n- openstack-tox-docs http://logs.openstack.org/57/626057/4/check/openstack-tox-docs/cc11966/html/ : SUCCESS in 5m 57s\n- cyborg-devstack http://logs.openstack.org/57/626057/4/check/cyborg-devstack/7f0d795/ : FAILURE in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"37d1359fe31bfba921a71eb7b9c819a83a6f22cf","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-12 07:31:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d36d14175b027ae1302c4dc46c83e234a7c5ddbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 07:51:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/5/check/openstack-tox-pep8/b09ac07/ : FAILURE in 7m 49s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/5/check/openstack-tox-py27/7a35b83/ : FAILURE in 7m 35s\n- openstack-tox-docs http://logs.openstack.org/57/626057/5/check/openstack-tox-docs/0136644/html/ : SUCCESS in 6m 08s\n- cyborg-devstack http://logs.openstack.org/57/626057/5/check/cyborg-devstack/b4614b6/ : FAILURE in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"78d9685a5ffdfdab19a2b7a98db889e49c71ff71","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-12 07:51:30.000000000","message":"Topic changed from 627427 to cyborg-nova-pilot","accounts_in_message":[],"_revision_number":5},{"id":"6223a5c87932f7a74802f59711e9a23c886a1466","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-13 22:55:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8d6b089192afc8a4c3d77e09f87092dab3aa1e40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 23:08:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/6/check/openstack-tox-pep8/6c6e2ae/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/6/check/openstack-tox-py27/ebe0892/ : FAILURE in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/57/626057/6/check/openstack-tox-docs/8ebef0a/html/ : SUCCESS in 4m 46s\n- cyborg-devstack http://logs.openstack.org/57/626057/6/check/cyborg-devstack/66af22d/ : FAILURE in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9f68cc1e2146cfe52b197bd6af442b66e4eb5c39","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-14 05:31:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9ebdebc21fc4154f4531c7e38fbe568bfe75d846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 05:43:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/7/check/openstack-tox-pep8/1550137/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/7/check/openstack-tox-py27/b49430d/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/57/626057/7/check/openstack-tox-docs/a03335c/html/ : SUCCESS in 4m 42s\n- cyborg-devstack http://logs.openstack.org/57/626057/7/check/cyborg-devstack/023a646/ : FAILURE in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5d78daba0c5161df2345046a420bc96b4e381d55","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-04-16 09:52:20.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nThanks for your efforts Sundar, please see my comments inline.","accounts_in_message":[],"_revision_number":7},{"id":"d0c4727eed80bc1e2db27da3d16bdb35238abe3d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-17 01:11:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d6a65d7bff65181d652287dfa8cb43edce48e3a1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-17 01:15:19.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nThanks, Xinran.","accounts_in_message":[],"_revision_number":7},{"id":"73e8f0a5b25aab27f4237171c833e616634538fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 01:30:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/8/check/openstack-tox-pep8/83980ee/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/8/check/openstack-tox-py27/48975bd/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/57/626057/8/check/openstack-tox-docs/6f26437/html/ : SUCCESS in 4m 22s\n- cyborg-devstack http://logs.openstack.org/57/626057/8/check/cyborg-devstack/538c5dc/ : FAILURE in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9179ed41763a99aa17aa2c8e587da4159476a0df","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-18 08:37:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ae20a74d3c4ac0764e443b3378db8e2f8a7046b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 08:59:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/9/check/openstack-tox-pep8/8c8ea49/ : FAILURE in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/9/check/openstack-tox-py27/32de96f/ : FAILURE in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/57/626057/9/check/openstack-tox-docs/6c68a70/html/ : SUCCESS in 6m 28s\n- cyborg-devstack http://logs.openstack.org/57/626057/9/check/cyborg-devstack/9f49e20/ : FAILURE in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"93c347c305bcaed3acaf62b3aef0683504f51f7a","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-04-24 08:05:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"18e8ad3e9c8c04456dfd1b049cfe3206e00b5bb0","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-24 23:46:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"51d64bc07eb7bbc1a01a39d6c03d34d69d6c2090","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-04-25 02:55:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"033a520920546fa01a50d70889fec26761ce0df0","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-25 15:01:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"35c4df9d842ebeffdd0bec4f94dbaa29cbd668a6","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-04-26 02:45:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"079cbd2012aae4d7a3770e2fcfffa698053ae795","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-11 07:54:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"eeddf3898b9d00a87f653ef29fa831bd7b038771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 08:12:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/10/check/openstack-tox-pep8/641ae0d/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/10/check/openstack-tox-py27/b683b2a/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/57/626057/10/check/openstack-tox-docs/5e5832c/html/ : SUCCESS in 4m 05s\n- cyborg-devstack http://logs.openstack.org/57/626057/10/check/cyborg-devstack/b31ab29/ : FAILURE in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9f8731107f1fc41e8bea79b01173a376c29049b3","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-14 10:44:52.000000000","message":"Patch Set 10: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"31db587bf4ec4dd962b3286a9e6ddd817192af88","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-14 10:50:41.000000000","message":"Patch Set 10:\n\nThanks for your work, Sundar. I think It\u0027s really a little difficult to review such a big patch.\n\nIf you want to split, I can review after split. If you still think a big one is OKay, I will also review in deep in next time. :)","accounts_in_message":[],"_revision_number":10},{"id":"228a8373a84b63b533fc160af0588c5ceb34cb53","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-14 11:43:45.000000000","message":"Patch Set 10:\n\nYikun, Cyborg community asked me to merge my smaller patches, and even voted unanimously on that!","accounts_in_message":[],"_revision_number":10},{"id":"7dfda184581e70645d1534b74ba9b8ef32aec5d8","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-14 11:51:54.000000000","message":"Patch Set 10:\n\n\u003e Yikun, Cyborg community asked me to merge my smaller patches, and\n \u003e even voted unanimously on that!\n\nI know, it\u0027s better for poc demo show and let people understand what you do by only one patch.\n\nSo, It\u0027s just a suggestion (or to say a question). I said in my last comment, I will also review in deep if you are not going to re-org your code.","accounts_in_message":[],"_revision_number":10},{"id":"ff7ce58dac30cde964463eeef313f15c4f851686","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-05-15 02:34:27.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)\n\nHi Sundar,\nPlz look at my comment in line.","accounts_in_message":[],"_revision_number":10},{"id":"07c6d73fdccb0b7e02331e0534314805055064df","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2019-05-20 11:35:31.000000000","message":"Patch Set 10: Code-Review-1\n\n(10 comments)\n\nPlease see my comments inline.","accounts_in_message":[],"_revision_number":10},{"id":"754c1781369a407be856aac16208de0696d0dbc6","author":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"date":"2019-05-28 11:38:03.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)\n\nPlease see my comments and questions.","accounts_in_message":[],"_revision_number":10},{"id":"bf3a68c47fcd96ad8073501e99951855243d7e81","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-06-03 07:21:59.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)\n\nHi Sundar, thanks a lot for your great job! Please see my comment inline :)","accounts_in_message":[],"_revision_number":10},{"id":"4607762d09182c08248a35df9a2a6514eb67cffc","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-06-12 08:30:35.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nJust encounter a bug, please see the comment inline.","accounts_in_message":[],"_revision_number":10},{"id":"ddf8f15a0db05cb25c377d9181f173a2448f08cc","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-19 04:42:42.000000000","message":"Patch Set 10:\n\n(34 comments)\n\nThanks for the comments.","accounts_in_message":[],"_revision_number":10},{"id":"52127971ea3681051e2ddd51d6091318e72ff742","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-19 21:53:57.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"6aec8549a6d0fb04e460aabaf48a8af9d0d13386","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-20 04:49:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6be229a80939bedba6fed894b4b2d5c948c70aa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 05:07:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/11/check/openstack-tox-pep8/fa0a203/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/11/check/openstack-tox-py27/dd8ca54/ : SUCCESS in 6m 31s\n- openstack-tox-docs http://logs.openstack.org/57/626057/11/check/openstack-tox-docs/b40bf16/html/ : SUCCESS in 6m 51s\n- cyborg-devstack http://logs.openstack.org/57/626057/11/check/cyborg-devstack/f12d14e/ : FAILURE in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"955a194aad7c504bdcffb6f949fa41550f15855b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-20 05:27:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1dd53fc98c7536d795cefffa513f5902197ce11b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-20 05:31:48.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2ffdc5118e8b0c204b2e66a08f446ff91f0c9514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 05:45:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/12/check/openstack-tox-pep8/994c603/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/12/check/openstack-tox-py27/2156bb9/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/57/626057/12/check/openstack-tox-docs/e4f72ae/html/ : SUCCESS in 4m 29s\n- cyborg-devstack http://logs.openstack.org/57/626057/12/check/cyborg-devstack/747c753/ : FAILURE in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"35b4ef2ecfdfe830b8276633803b26e3b8a94407","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-06-28 03:30:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)\n\nThanks Sundar, please see my comments inline.","accounts_in_message":[],"_revision_number":12},{"id":"bc7f1c76339cdd0a3975bc15ef84080198e60a40","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-01 03:28:56.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b07d65415fd850996fc8051e6c6f01561b604aeb","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-01 05:34:54.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d7a00350e07ebb5d97f9e79844c0359b0af7dc0f","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-03 02:31:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"423cf3702d2c4fb3dc56ba774e222688ac2d32b1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-08 00:22:53.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ae15cb994e0952f442c8d144efd3c9a614f4a31a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 00:41:10.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/13/check/openstack-tox-pep8/ab0b0f8/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/13/check/openstack-tox-py27/107c95d/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/57/626057/13/check/openstack-tox-docs/ec154a1/html/ : SUCCESS in 4m 11s\n- cyborg-devstack http://logs.openstack.org/57/626057/13/check/cyborg-devstack/872abe3/ : FAILURE in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bbf403adc89edc8a77a5eb2e5826a2c7074747e2","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-10 08:56:08.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"44acdcebd33d303501743b0d4746e60870205da9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 09:22:56.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/14/check/openstack-tox-pep8/d549d38/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/14/check/openstack-tox-py27/8ee7a5d/ : FAILURE in 2m 45s\n- openstack-tox-docs http://logs.openstack.org/57/626057/14/check/openstack-tox-docs/2866b17/html/ : SUCCESS in 3m 27s\n- cyborg-devstack http://logs.openstack.org/57/626057/14/check/cyborg-devstack/ce9e71a/ : FAILURE in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"32a44fbeaf10d66da220c3536fcefeaf95bcf555","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 04:46:29.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"83dbfa65f5c9e33acc6b9b6f6db6377db340ca5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 05:21:40.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/15/check/openstack-tox-pep8/75d810b/ : FAILURE in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/15/check/openstack-tox-py27/780defa/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/57/626057/15/check/openstack-tox-docs/36c19a9/html/ : SUCCESS in 3m 59s\n- cyborg-devstack http://logs.openstack.org/57/626057/15/check/cyborg-devstack/258dd33/ : SUCCESS in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a6340755cdb2b27cf67ee5a6510c9b6eedd32062","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 05:49:02.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"32ad1502313840714b2f9c7b8f4cd735fa73c701","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 06:26:42.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/57/626057/16/check/openstack-tox-pep8/75d75fa/ : FAILURE in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/57/626057/16/check/openstack-tox-py27/3eaadb3/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/57/626057/16/check/openstack-tox-docs/53f6c98/html/ : SUCCESS in 3m 54s\n- cyborg-devstack http://logs.openstack.org/57/626057/16/check/cyborg-devstack/a22666b/ : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b28755e3236bda761c9a10c3557b06007babe475","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-07-29 19:33:26.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"c2008a3a38c84f4703b441f33aa7790124950047","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-29 23:50:44.000000000","message":"Patch Set 16: Workflow-1","accounts_in_message":[],"_revision_number":16},{"id":"ef1554eddc9dbb767038c84832f0dad5fa319e9a","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-09-24 06:25:48.000000000","message":"Patch Set 16:\n\nmaybe we can abandon these deprecated patches now?","accounts_in_message":[],"_revision_number":16},{"id":"23b6b15489764cf04a221dfbcda4a4a27177efc3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-01-29 08:00:18.000000000","message":"Abandoned\n\ndeprecated","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"f3b4b306b8067084f7f0e99f1954b335bbc353f5","revisions":{"73bd9149dc498c8080e8fafc5f2e21672fb371c9":{"kind":"REWORK","_number":1,"created":"2018-12-19 00:24:58.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/1"}}},"commit":{"parents":[{"commit":"a1af48107b46c16a55c192eb3c3b3b4da6763dcc","subject":"Initial checkin, with only README.rst.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a1af48107b46c16a55c192eb3c3b3b4da6763dcc"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:23:51.000000000","tz":-480},"subject":"Disable old tests so that \u0027tox -epy27\u0027 passes.","message":"Disable old tests so that \u0027tox -epy27\u0027 passes.\n\n\u0027tox -epep8\u0027 also passes at this point.\n\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/73bd9149dc498c8080e8fafc5f2e21672fb371c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/73bd9149dc498c8080e8fafc5f2e21672fb371c9"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"df79e45c4d6183a1664d5df24189511fbde27a27":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-12-20 01:29:55.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/2"}}},"commit":{"parents":[{"commit":"8f0bd275b42c806f46ca92821b78c785a5ff94b2","subject":"Initial checkin, with only README.rst.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8f0bd275b42c806f46ca92821b78c785a5ff94b2"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-20 01:51:59.000000000","tz":-480},"subject":"Disable old tests so that \u0027tox -epy27\u0027 passes.","message":"Disable old tests so that \u0027tox -epy27\u0027 passes.\n\n\u0027tox -epep8\u0027 also passes at this point.\n\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/df79e45c4d6183a1664d5df24189511fbde27a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/df79e45c4d6183a1664d5df24189511fbde27a27"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"296d9f3f696d18ef3f8ec5945b86287c9c68024b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-21 04:41:10.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/3"}}},"commit":{"parents":[{"commit":"316ac0aec97335bb7b357b4d5d4c220b33c0cd2d","subject":"Initial checkin, with only README.rst.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/316ac0aec97335bb7b357b4d5d4c220b33c0cd2d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-03-21 05:20:05.000000000","tz":-420},"subject":"Disable old tests so that \u0027tox -epy27\u0027 passes.","message":"Disable old tests so that \u0027tox -epy27\u0027 passes.\n\n\u0027tox -epep8\u0027 also passes at this point.\n\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/296d9f3f696d18ef3f8ec5945b86287c9c68024b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/296d9f3f696d18ef3f8ec5945b86287c9c68024b"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"a2fb38280f632734a19ea16727cf72c0e7ade5eb":{"kind":"REWORK","_number":4,"created":"2019-04-03 22:36:38.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/4"}}},"commit":{"parents":[{"commit":"6cacf878da836393450e125179135040b1b9dc97","subject":"Initial checkin, with only README.rst.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6cacf878da836393450e125179135040b1b9dc97"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-03 17:12:47.000000000","tz":-420},"subject":"Disable old tests so that \u0027tox -epy27\u0027 passes.","message":"Disable old tests so that \u0027tox -epy27\u0027 passes.\n\n\u0027tox -epep8\u0027 also passes at this point.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a2fb38280f632734a19ea16727cf72c0e7ade5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a2fb38280f632734a19ea16727cf72c0e7ade5eb"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"1a62e41a6ca2549201ee9c4f43d34987be60b726":{"kind":"REWORK","_number":5,"created":"2019-04-12 07:31:41.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/5"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-12 07:20:27.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1a62e41a6ca2549201ee9c4f43d34987be60b726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1a62e41a6ca2549201ee9c4f43d34987be60b726"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"a18f472f51bec84350c93e4cf841de7ea7736477":{"kind":"REWORK","_number":6,"created":"2019-04-13 22:55:32.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/6"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-13 22:45:16.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a18f472f51bec84350c93e4cf841de7ea7736477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a18f472f51bec84350c93e4cf841de7ea7736477"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"0de0fea5e24db5f098ca89aa8a574e53f2321bd3":{"kind":"REWORK","_number":7,"created":"2019-04-14 05:31:13.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/7"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-14 05:21:58.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0de0fea5e24db5f098ca89aa8a574e53f2321bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0de0fea5e24db5f098ca89aa8a574e53f2321bd3"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"92281b6171929b7deca9fe226b7db792313bf8dc":{"kind":"REWORK","_number":8,"created":"2019-04-17 01:11:33.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/8"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-17 01:08:01.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/92281b6171929b7deca9fe226b7db792313bf8dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/92281b6171929b7deca9fe226b7db792313bf8dc"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"92a19909dd7a95deb2fae5b32bf63081d9c18e26":{"kind":"REWORK","_number":9,"created":"2019-04-18 08:37:50.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/9"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-18 08:35:40.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/92a19909dd7a95deb2fae5b32bf63081d9c18e26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/92a19909dd7a95deb2fae5b32bf63081d9c18e26"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"bc25e9b1b2cd4d34b195184d211acd29c6159dc6":{"kind":"REWORK","_number":10,"created":"2019-05-11 07:54:28.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/10"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-11 07:53:15.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bc25e9b1b2cd4d34b195184d211acd29c6159dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bc25e9b1b2cd4d34b195184d211acd29c6159dc6"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"961737b9a1c544b5b56060fe067d6a711cde422b":{"kind":"REWORK","_number":11,"created":"2019-06-20 04:49:46.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/11"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-06-20 04:49:15.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/961737b9a1c544b5b56060fe067d6a711cde422b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/961737b9a1c544b5b56060fe067d6a711cde422b"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"2c839f65e2b1f83168dca4770fbc7474263261f3":{"kind":"REWORK","_number":12,"created":"2019-06-20 05:27:38.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/12"}}},"commit":{"parents":[{"commit":"d8384d62efa450ad396620282b902a49d1b8505d","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8384d62efa450ad396620282b902a49d1b8505d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-06-20 05:26:48.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c839f65e2b1f83168dca4770fbc7474263261f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c839f65e2b1f83168dca4770fbc7474263261f3"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"5b657318a086842a1412c93290ae898337519b59":{"kind":"REWORK","_number":13,"created":"2019-07-08 00:22:53.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/13"}}},"commit":{"parents":[{"commit":"8528da32404aef60e48f1c899fa8af8a0e9bdbee","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8528da32404aef60e48f1c899fa8af8a0e9bdbee"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-08 00:20:16.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b657318a086842a1412c93290ae898337519b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b657318a086842a1412c93290ae898337519b59"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"813124970541eb06eb19d81fae4df42beddae9f8":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-07-10 08:56:08.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/14"}}},"commit":{"parents":[{"commit":"d91de7c2eb201f2b5c0951b1ab5dd26786d9ed7e","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d91de7c2eb201f2b5c0951b1ab5dd26786d9ed7e"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-10 08:53:57.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/813124970541eb06eb19d81fae4df42beddae9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/813124970541eb06eb19d81fae4df42beddae9f8"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"198fc107966b96e4711ac249f50ecab91f9abef1":{"kind":"REWORK","_number":15,"created":"2019-07-12 04:46:29.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/15"}}},"commit":{"parents":[{"commit":"d91de7c2eb201f2b5c0951b1ab5dd26786d9ed7e","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d91de7c2eb201f2b5c0951b1ab5dd26786d9ed7e"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 04:43:49.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/198fc107966b96e4711ac249f50ecab91f9abef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/198fc107966b96e4711ac249f50ecab91f9abef1"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"},"f3b4b306b8067084f7f0e99f1954b335bbc353f5":{"kind":"REWORK","_number":16,"created":"2019-07-12 05:49:02.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/57/626057/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/57/626057/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/57/626057/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/57/626057/16"}}},"commit":{"parents":[{"commit":"d91de7c2eb201f2b5c0951b1ab5dd26786d9ed7e","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into merge-branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d91de7c2eb201f2b5c0951b1ab5dd26786d9ed7e"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-12-19 01:13:49.000000000","tz":-480},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 05:15:02.000000000","tz":-420},"subject":"Changes for v2 API and Nova integration for VM creation.","message":"Changes for v2 API and Nova integration for VM creation.\n\nCyborg now supports both AFaaS and DaaS use cases.\n\nCyborg now queries Glance for function ID using openstack-sdk.\nSo, /etc/opentack/clouds.yaml must be set up. That is done by\ndefault in devstack. (Bitstream download in the agent still\nuses glance client.)\n\nAdded Placement client based on openstack-sdk. Traits get\nupdated after successful programming.\n\nTopic: cyborg-nova-pilot\nChange-Id: Icb92206b4f6bcebc4c0be219d3937a1f211cac82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f3b4b306b8067084f7f0e99f1954b335bbc353f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f3b4b306b8067084f7f0e99f1954b335bbc353f5"}]},"branch":"refs/heads/feature/cyborg-nova-pilot"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
