)]}'
{"id":"openstack%2Fcyborg~657743","triplet_id":"openstack%2Fcyborg~master~I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5","project":"openstack/cyborg","branch":"master","topic":"driver","hashtags":[],"change_id":"I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5","subject":"Improve Intel sysinfo based FPGA driver","status":"MERGED","created":"2019-05-08 07:22:50.000000000","updated":"2019-05-20 02:46:16.000000000","submitted":"2019-05-20 02:46:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657743-1558320376546-334d6ac5","meta_rev_id":"ffdae70fe42b37b6bacfcd23dff49a10815e7263","_number":657743,"virtual_id_number":657743,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"date":"2019-05-20 02:04:36.000000000","_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-05-20 02:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":2,"date":"2019-05-20 02:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":2,"date":"2019-05-19 05:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":-1,"date":"2019-05-20 02:39:06.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":1,"date":"2019-05-20 02:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 08:22:45.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2019-05-09 08:22:48.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2019-05-19 05:37:49.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-05-20 02:37:40.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-05-20 02:39:06.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-05-20 02:46:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc74b325c43ec29d0f2539165cdd1422b00cd77a","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-08 07:22:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb23d551fc08d795d84f52689a0315e41beccc63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 07:58:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/43/657743/1/check/openstack-tox-pep8/8e2f87b/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/43/657743/1/check/openstack-tox-py27/0fd5efc/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/43/657743/1/check/openstack-tox-docs/5d77490/html/ : SUCCESS in 6m 09s\n- cyborg-devstack http://logs.openstack.org/43/657743/1/check/cyborg-devstack/d8553ab/ : SUCCESS in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3158e129f3445151b5a9ad8675b953ba416c7167","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-09 08:22:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1a5666e91ccc684aacbb1fd79c1881039604072c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:05:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/43/657743/2/check/openstack-tox-pep8/2dbe80e/ : FAILURE in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/43/657743/2/check/openstack-tox-py27/5878aa6/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/43/657743/2/check/openstack-tox-docs/581da68/html/ : SUCCESS in 3m 56s\n- cyborg-devstack http://logs.openstack.org/43/657743/2/check/cyborg-devstack/d30e956/ : SUCCESS in 41m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ee0c8fe6086d8b961f62d0e6b7c9e231f273bbf","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-09 10:11:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"12c195474448ee7580f2daba7d938f1fdc74d309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 10:46:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/43/657743/3/check/openstack-tox-pep8/765d8da/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/43/657743/3/check/openstack-tox-py27/d9a2d40/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/43/657743/3/check/openstack-tox-docs/bef82cb/html/ : SUCCESS in 4m 58s\n- cyborg-devstack http://logs.openstack.org/43/657743/3/check/cyborg-devstack/b577629/ : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"abcbc81d83a8102adaa31774dcdad619af3802f2","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-15 08:13:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a4e4b7fdd54e3b77aacaaac0bbfe025e823b5bf7","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-15 08:44:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d3e9a963550407ef651086f4e76fff572de0d1b3","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-15 08:47:59.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7f288a6c77ba62f14f3d48f81daf6dd2154cd4e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 09:24:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/43/657743/6/check/openstack-tox-pep8/9a32a30/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/43/657743/6/check/openstack-tox-py27/a144a65/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/43/657743/6/check/openstack-tox-docs/08ddb5c/html/ : SUCCESS in 4m 09s\n- cyborg-devstack http://logs.openstack.org/43/657743/6/check/cyborg-devstack/05c9846/ : SUCCESS in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8564f3ea588889a68576a453d9595ce96bca6048","author":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"date":"2019-05-19 05:37:49.000000000","message":"Patch Set 6: Code-Review+2\n\nrc and traits are stored in attribute db?","accounts_in_message":[],"_revision_number":6},{"id":"5d67e506090a87cdf6fccd6abae02c50f4e83537","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-05-20 02:04:36.000000000","message":"Patch Set 6:\n\n\u003e rc and traits are stored in attribute db?\n\nyes, in current design.","accounts_in_message":[],"_revision_number":6},{"id":"fe190900c291662aff9b9a7cea482d87ac7d1ad0","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-20 02:37:40.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a361f75ab158af2c6db4e10ad811d63dfd10db5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 02:37:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7d33d304676f398b033d88a4a8e23d7d526feb95","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-20 02:38:07.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks xinran to improve the driver.","accounts_in_message":[],"_revision_number":6},{"id":"0122880e38964c8103be3a1fe837dbf3cb5bc934","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-05-20 02:39:06.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"99d358f420ac5ac2ea0da68b468f6a63e7300b2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 02:46:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9b7d342ac8623d2291037508bef35304a491de9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 02:46:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/43/657743/6/gate/openstack-tox-pep8/a037cdd/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/43/657743/6/gate/openstack-tox-py27/16e2eeb/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/43/657743/6/gate/openstack-tox-docs/b064604/html/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0bc8ca656288a49bd217f6a1515ba68ff767b737","revisions":{"d99cc3f55ed2449cdf431933cd309532a309506c":{"kind":"REWORK","_number":1,"created":"2019-05-08 07:22:50.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/43/657743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/43/657743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/43/657743/1"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Ubuntu","email":"xin-ran.wang@intel.com","date":"2019-05-08 07:21:59.000000000","tz":0},"committer":{"name":"Ubuntu","email":"xin-ran.wang@intel.com","date":"2019-05-08 07:21:59.000000000","tz":0},"subject":"Improve Intel sysinfo based FPGA driver","message":"Improve Intel sysinfo based FPGA driver\n\nEach driver should report the resource class and traits to Cyborg agent in\norder to let agent pass them to conductor, then conductor could call placment\nto update them.\n\nIn FPGA case, driver should discover RC as \"CUSTOM_ACCELERATOR_FPGA\" and also\ndiscover the function id, region type if exist, as TRAITS.\n\nThe placement report work depends on this patch.\n\nWe should also improve the OPEA in the future.\n\nChange-Id: I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d99cc3f55ed2449cdf431933cd309532a309506c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d99cc3f55ed2449cdf431933cd309532a309506c"}]},"branch":"refs/heads/master"},"ee7a11b6a8376bd56ac4768e80bd44cf59354432":{"kind":"REWORK","_number":2,"created":"2019-05-09 08:22:19.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/43/657743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/43/657743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/43/657743/2"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Ubuntu","email":"xin-ran.wang@intel.com","date":"2019-05-08 07:21:59.000000000","tz":0},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2019-05-09 08:22:12.000000000","tz":0},"subject":"Improve Intel sysinfo based FPGA driver","message":"Improve Intel sysinfo based FPGA driver\n\nEach driver should report the resource class and traits to Cyborg agent in\norder to let agent pass them to conductor, then conductor could call placment\nto update them.\n\nIn FPGA case, driver should discover RC as \"CUSTOM_ACCELERATOR_FPGA\" and also\ndiscover the function id, region type if exist, as TRAITS.\n\nThe placement report work depends on this patch.\n\nWe should also improve the OPEA in the future.\n\nChange-Id: I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ee7a11b6a8376bd56ac4768e80bd44cf59354432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ee7a11b6a8376bd56ac4768e80bd44cf59354432"}]},"branch":"refs/heads/master"},"5f459966045d2d04ec467b736096ab6b2422d189":{"kind":"REWORK","_number":3,"created":"2019-05-09 10:11:08.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/43/657743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/43/657743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/43/657743/3"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Ubuntu","email":"xin-ran.wang@intel.com","date":"2019-05-08 07:21:59.000000000","tz":0},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2019-05-09 10:11:00.000000000","tz":0},"subject":"Improve Intel sysinfo based FPGA driver","message":"Improve Intel sysinfo based FPGA driver\n\nEach driver should report the resource class and traits to Cyborg agent in\norder to let agent pass them to conductor, then conductor could call placment\nto update them.\n\nIn FPGA case, driver should discover RC as \"CUSTOM_ACCELERATOR_FPGA\" and also\ndiscover the function id, region type if exist, as TRAITS.\n\nThe placement report work depends on this patch.\n\nWe should also improve the OPEA in the future.\n\nChange-Id: I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5f459966045d2d04ec467b736096ab6b2422d189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5f459966045d2d04ec467b736096ab6b2422d189"}]},"branch":"refs/heads/master"},"6a37e7ed8be323fa21eb5d49dcfcda2d96d49555":{"kind":"REWORK","_number":4,"created":"2019-05-15 08:13:35.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/43/657743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/43/657743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/43/657743/4"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-05-07 10:19:22.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-05-07 10:19:22.000000000","tz":480},"subject":"Improve Intel sysinfo based FPGA driver","message":"Improve Intel sysinfo based FPGA driver\n\nEach driver should report the resource class and traits to Cyborg agent in\norder to let agent pass them to conductor, then conductor could call placment\nto update them.\n\nIn FPGA case, driver should discover RC as \"CUSTOM_ACCELERATOR_FPGA\" and also\ndiscover the function id, region type if exist, as TRAITS.\n\nThe placement report work depends on this patch.\n\nWe should also improve the OPEA in the future.\n\nChange-Id: I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6a37e7ed8be323fa21eb5d49dcfcda2d96d49555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6a37e7ed8be323fa21eb5d49dcfcda2d96d49555"}]},"branch":"refs/heads/master"},"ce845ed63691406d06b9e72e3b13716f4c0ad0c6":{"kind":"REWORK","_number":5,"created":"2019-05-15 08:44:45.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/43/657743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/43/657743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/43/657743/5"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-05-07 10:19:22.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-05-15 08:44:18.000000000","tz":480},"subject":"Improve Intel sysinfo based FPGA driver","message":"Improve Intel sysinfo based FPGA driver\n\nEach driver should report the resource class and traits to Cyborg agent in\norder to let agent pass them to conductor, then conductor could call placment\nto update them.\n\nIn FPGA case, driver should discover RC as \"CUSTOM_ACCELERATOR_FPGA\" and also\ndiscover the function id, region type if exist, as TRAITS.\n\nThe placement report work depends on this patch.\n\nWe should also improve the OPEA in the future.\n\nChange-Id: I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ce845ed63691406d06b9e72e3b13716f4c0ad0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ce845ed63691406d06b9e72e3b13716f4c0ad0c6"}]},"branch":"refs/heads/master"},"0bc8ca656288a49bd217f6a1515ba68ff767b737":{"kind":"REWORK","_number":6,"created":"2019-05-15 08:47:59.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/43/657743/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/43/657743/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/43/657743/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/43/657743/6"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-05-07 10:19:22.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2019-05-15 08:46:47.000000000","tz":0},"subject":"Improve Intel sysinfo based FPGA driver","message":"Improve Intel sysinfo based FPGA driver\n\nEach driver should report the resource class and traits to Cyborg agent in\norder to let agent pass them to conductor, then conductor could call placment\nto update them.\n\nIn FPGA case, driver should discover RC as \"CUSTOM_ACCELERATOR_FPGA\" and also\ndiscover the function id, region type if exist, as TRAITS.\n\nThe placement report work depends on this patch.\n\nWe should also improve the OPEA in the future.\n\nChange-Id: I4a679a7b93de62dbca25e293f7c2e7e71ffad2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0bc8ca656288a49bd217f6a1515ba68ff767b737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0bc8ca656288a49bd217f6a1515ba68ff767b737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
