)]}'
{"id":"openstack%2Fcyborg~659292","triplet_id":"openstack%2Fcyborg~master~Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f","subject":"bug fix: deploy every cyborg components correctly","status":"MERGED","created":"2019-05-15 13:20:02.000000000","updated":"2019-07-11 22:56:56.000000000","submitted":"2019-07-11 22:56:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659292-1562885816827-e25996b4","meta_rev_id":"fbc076067898ebe1264bb757f126866714b2a455","_number":659292,"virtual_id_number":659292,"owner":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-07-11 22:56:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":2,"date":"2019-07-11 22:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":1,"date":"2019-07-11 22:46:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 16:12:19.000000000","updated_by":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"reviewer":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"state":"REVIEWER"},{"updated":"2019-07-11 22:46:44.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-07-11 22:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bce6c8a4f9d4c562f559e3976bec8718b5c4a32","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-15 13:20:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b4d265c5e8fd4f373a938110fc20b91860066a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:01:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/659292/1/check/openstack-tox-pep8/7e9449b/ : FAILURE in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/92/659292/1/check/openstack-tox-py27/27a88d0/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/92/659292/1/check/openstack-tox-docs/6b4d61d/ : FAILURE in 4m 49s\n- cyborg-devstack http://logs.openstack.org/92/659292/1/check/cyborg-devstack/536cdfa/ : SUCCESS in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc1a9709d0d4e9e44f3eb22852a5a4f839a28a7f","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-15 23:41:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nThanks, Shaohe.","accounts_in_message":[],"_revision_number":1},{"id":"7746217d938c9e5b1bc01db55d29681189e85f0f","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-16 05:53:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"05ae8bdee2b66c66efe648a74d0ecb6be836f8fa","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-16 09:37:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d8536558fd7057c57f4c8a34a49258685c4f918","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:12:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/659292/2/check/openstack-tox-pep8/800dd31/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/92/659292/2/check/openstack-tox-py27/babd491/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/92/659292/2/check/openstack-tox-docs/acc9614/html/ : SUCCESS in 4m 49s\n- cyborg-devstack http://logs.openstack.org/92/659292/2/check/cyborg-devstack/604593d/ : SUCCESS in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44838568489739cdaa672c120f26c3ba889a7652","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-16 23:15:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nThanks, Shaohe. This is close now.","accounts_in_message":[],"_revision_number":2},{"id":"21b1dbd85b08f2f332a34029b9ccc848a9b69d64","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-20 06:09:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8653b3cd63505bd12ad022ff86759fb102cffaab","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-20 06:14:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3352a8e1a2b2856a66db455509e7055edc704fb3","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-05-20 06:15:21.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"006696bb5d765c9cad74ec6dc3fe8608d38a756d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 06:50:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/659292/4/check/openstack-tox-pep8/bbce82f/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/92/659292/4/check/openstack-tox-py27/12e0ed7/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/92/659292/4/check/openstack-tox-docs/4757e79/html/ : SUCCESS in 4m 15s\n- cyborg-devstack http://logs.openstack.org/92/659292/4/check/cyborg-devstack/0039402/ : SUCCESS in 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9475a0183d01497806fa33838e73428ec5916c11","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-20 22:23:30.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nThanks, Shaohe.","accounts_in_message":[],"_revision_number":4},{"id":"c3fd32da97248d3384a514239477c9a9e273f442","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-05-21 16:02:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11ce017393c68aae64b8947233546141f83b71f0","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-05-21 16:12:19.000000000","message":"Patch Set 4:\n\nShaohe, I would just remove the commented sections and I can work to get this added to the devstack multi-host guide.","accounts_in_message":[],"_revision_number":4},{"id":"02dc425e1d1507bcc7d761e347f7de6114d86c3f","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-06-12 12:23:28.000000000","message":"Patch Set 4:\n\nDonny Davis.\nThank you, let\u0027s remove this section.","accounts_in_message":[],"_revision_number":4},{"id":"d9b0af79bbee956a495d1730d5b7c8527f3533a4","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-06-19 03:41:24.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2a5dfd2703c13a798bcb40205d9cda5104990e77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 04:28:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/659292/5/check/openstack-tox-pep8/b948882/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/92/659292/5/check/openstack-tox-py27/03f8682/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/92/659292/5/check/openstack-tox-docs/aa0f529/html/ : SUCCESS in 4m 55s\n- cyborg-devstack http://logs.openstack.org/92/659292/5/check/cyborg-devstack/a657eff/ : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6fef64aca6fff22003813ee0896c4f1a35756ca6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-19 18:06:24.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM. Giving time for others to review before adding +2.","accounts_in_message":[],"_revision_number":5},{"id":"0f99852906bd07683ebab99d84f52faec55a3d6d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-11 22:46:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d8a1fe66c07feac75c4ad7917326dcd730c9cb75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 22:46:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a85d4a3add91173f618430f237e1d6f704e14fa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 22:56:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fbc076067898ebe1264bb757f126866714b2a455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 22:56:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/659292/5/gate/openstack-tox-pep8/04121d5/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/92/659292/5/gate/openstack-tox-py27/a8bd65a/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/92/659292/5/gate/openstack-tox-docs/83c7dcc/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"34f8c6a1479c2854577d83fe93a92ca647bbbd97","revisions":{"661aa8e70875f70880c48e533ca8d37d33145105":{"kind":"REWORK","_number":1,"created":"2019-05-15 13:20:02.000000000","uploader":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"ref":"refs/changes/92/659292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/92/659292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/92/659292/1"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-15 12:43:12.000000000","tz":0},"committer":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-15 13:19:47.000000000","tz":0},"subject":"bug fix: deploy every cyborg components correctly","message":"bug fix: deploy every cyborg components correctly\n\nDifferent components need different config.\n\nChange-Id: Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/661aa8e70875f70880c48e533ca8d37d33145105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/661aa8e70875f70880c48e533ca8d37d33145105"}]},"branch":"refs/heads/master"},"555a7eca60cd7af39f005217fb43d9f23c569e6e":{"kind":"REWORK","_number":2,"created":"2019-05-16 09:37:45.000000000","uploader":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"ref":"refs/changes/92/659292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/92/659292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/92/659292/2"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-15 12:43:12.000000000","tz":0},"committer":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-16 09:37:27.000000000","tz":0},"subject":"bug fix: deploy every cyborg components correctly","message":"bug fix: deploy every cyborg components correctly\n\nDifferent components need different config.\n\nChange-Id: Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/555a7eca60cd7af39f005217fb43d9f23c569e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/555a7eca60cd7af39f005217fb43d9f23c569e6e"}]},"branch":"refs/heads/master"},"ce293de569975dc215859a52e1f6fb1449898918":{"kind":"REWORK","_number":3,"created":"2019-05-20 06:14:05.000000000","uploader":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"ref":"refs/changes/92/659292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/92/659292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/92/659292/3"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-15 12:43:12.000000000","tz":0},"committer":{"name":"ShaoHe Feng","email":"shaohe.feng@intel.com","date":"2019-05-20 06:13:38.000000000","tz":0},"subject":"bug fix: deploy every cyborg components correctly","message":"bug fix: deploy every cyborg components correctly\n\nDifferent components need different config.\n\nChange-Id: Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ce293de569975dc215859a52e1f6fb1449898918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ce293de569975dc215859a52e1f6fb1449898918"}]},"branch":"refs/heads/master"},"6eec62ed5ad5305ccf2579f9eba9f13f9fd50a32":{"kind":"REWORK","_number":4,"created":"2019-05-20 06:15:21.000000000","uploader":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"ref":"refs/changes/92/659292/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/92/659292/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/92/659292/4"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-15 12:43:12.000000000","tz":0},"committer":{"name":"ShaoHe Feng","email":"shaohe.feng@intel.com","date":"2019-05-20 06:15:13.000000000","tz":0},"subject":"bug fix: deploy every cyborg components correctly","message":"bug fix: deploy every cyborg components correctly\n\nDifferent components need different config.\n\nChange-Id: Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6eec62ed5ad5305ccf2579f9eba9f13f9fd50a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6eec62ed5ad5305ccf2579f9eba9f13f9fd50a32"}]},"branch":"refs/heads/master"},"34f8c6a1479c2854577d83fe93a92ca647bbbd97":{"kind":"REWORK","_number":5,"created":"2019-06-19 03:41:24.000000000","uploader":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"ref":"refs/changes/92/659292/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/92/659292/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/92/659292/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/92/659292/5"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Shaohe Feng","email":"shaohe.feng@intel.com","date":"2019-05-15 12:43:12.000000000","tz":0},"committer":{"name":"ShaoHe Feng","email":"shaohe.feng@intel.com","date":"2019-06-19 03:41:12.000000000","tz":0},"subject":"bug fix: deploy every cyborg components correctly","message":"bug fix: deploy every cyborg components correctly\n\nDifferent components need different config.\n\nChange-Id: Id73cd0dd7eb2abc2d411cc3d5abbbd4e3df58f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/34f8c6a1479c2854577d83fe93a92ca647bbbd97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/34f8c6a1479c2854577d83fe93a92ca647bbbd97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
