)]}'
{"id":"openstack%2Fcyborg~660353","triplet_id":"openstack%2Fcyborg~master~Iecf1628060193e751233fc915be2c0b488e3e7b0","project":"openstack/cyborg","branch":"master","topic":"fix_gpu_re","hashtags":[],"change_id":"Iecf1628060193e751233fc915be2c0b488e3e7b0","subject":"Fix re defination of GPU","status":"MERGED","created":"2019-05-21 11:19:35.000000000","updated":"2019-07-10 03:48:15.000000000","submitted":"2019-07-10 03:48:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660353-1562730495602-0a1ee91c","meta_rev_id":"c19218d4a3e859822a96e9c3554b4ea6b69583ac","_number":660353,"virtual_id_number":660353,"owner":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-07-10 03:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":1,"date":"2019-07-03 09:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-07-03 09:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":2,"date":"2019-07-10 03:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":1,"date":"2019-07-10 03:39:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 11:20:21.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-06-05 03:43:26.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-06-11 10:05:39.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-10 03:39:44.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-07-10 03:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"536bb306f5e94c83bdb174ad4cc89c9e95495f87","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-21 11:19:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"502e64752512cfde2b2a5f2b062b442361833f74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/660353/1/check/openstack-tox-pep8/ff38bfb/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/53/660353/1/check/openstack-tox-py27/ff2c922/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/53/660353/1/check/openstack-tox-docs/dd46487/html/ : SUCCESS in 4m 25s\n- cyborg-devstack http://logs.openstack.org/53/660353/1/check/cyborg-devstack/db108ac/ : SUCCESS in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"285484f97beb8e39e281c5ee8c4f519d4fef2e68","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-05-22 01:14:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1571c000f586d2040fa11588f65fcf2842de6ff","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-22 02:59:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0cb71105e58d88bbf2f53e14807aa65cbc79139a","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-06-05 03:43:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9e2b65705a3b65016a132815b0bb6e430f9d558","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-11 10:05:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc3bd2d2f610413c9f3c82a7916db0612794e86b","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 09:18:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"011e6fca675fbab419dc649c74d3678717e370f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:08:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/660353/2/check/openstack-tox-pep8/5cb9c3a/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/53/660353/2/check/openstack-tox-py27/fa046ab/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/53/660353/2/check/openstack-tox-docs/3864678/html/ : SUCCESS in 3m 12s\n- cyborg-devstack http://logs.openstack.org/53/660353/2/check/cyborg-devstack/74b5b7d/ : SUCCESS in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e729c5a62bfcadc32f10d840820999ee8961e666","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-10 03:39:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"51d27a3a1bd23daf159e22160a6b5950e3941793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:40:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33cd1383f23fec730fbec61b2c00f60da3f43288","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:48:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/660353/2/gate/openstack-tox-pep8/aa89385/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/53/660353/2/gate/openstack-tox-py27/7c9f978/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/53/660353/2/gate/openstack-tox-docs/18d6d1e/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c41b5c77bb1265e2a10b9349b002573bf5dcf55d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:48:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3ed8db9749aaf851d9e64395bc6b6fdf1a1ccdf","revisions":{"eca8fd71415ca1a0c76e1afe2fa3dd90206a8e8e":{"kind":"REWORK","_number":1,"created":"2019-05-21 11:19:35.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/53/660353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/53/660353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/53/660353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/53/660353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/53/660353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/53/660353/1"}}},"commit":{"parents":[{"commit":"179bcd17115b8dab6dcee46a36d561bd807dcb5a","subject":"Fix auth config in \"Install Cyborg from Source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/179bcd17115b8dab6dcee46a36d561bd807dcb5a"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:19:15.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:19:15.000000000","tz":480},"subject":"Fix re defination of GPU","message":"Fix re defination of GPU\n\nThe bus/device/function could contain \"a-f\", but now the regex\ntemplate string is not cover this case.\n\nThis patch try to fix it.\n\nChange-Id: Iecf1628060193e751233fc915be2c0b488e3e7b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eca8fd71415ca1a0c76e1afe2fa3dd90206a8e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eca8fd71415ca1a0c76e1afe2fa3dd90206a8e8e"}]},"branch":"refs/heads/master"},"e3ed8db9749aaf851d9e64395bc6b6fdf1a1ccdf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-03 09:18:47.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/53/660353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/53/660353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/53/660353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/53/660353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/53/660353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/53/660353/2"}}},"commit":{"parents":[{"commit":"75ecc3ab85f49faad74bb8d3be1f0a9aa6d752c6","subject":"Update json module to jsonutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/75ecc3ab85f49faad74bb8d3be1f0a9aa6d752c6"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:19:15.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-03 09:18:47.000000000","tz":0},"subject":"Fix re defination of GPU","message":"Fix re defination of GPU\n\nThe bus/device/function could contain \"a-f\", but now the regex\ntemplate string is not cover this case.\n\nThis patch try to fix it.\n\nChange-Id: Iecf1628060193e751233fc915be2c0b488e3e7b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e3ed8db9749aaf851d9e64395bc6b6fdf1a1ccdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e3ed8db9749aaf851d9e64395bc6b6fdf1a1ccdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
