)]}'
{"id":"openstack%2Fcyborg~660548","triplet_id":"openstack%2Fcyborg~master~I4c7a62bba7faf081066446862b7718dee64bb7a1","project":"openstack/cyborg","branch":"master","topic":"ascend","hashtags":[],"change_id":"I4c7a62bba7faf081066446862b7718dee64bb7a1","subject":"Add AICHIP and GENERIC type in DB and obj","status":"MERGED","created":"2019-05-22 02:47:20.000000000","updated":"2019-07-09 13:58:34.000000000","submitted":"2019-07-09 13:58:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660548-1562680714409-31ef7618","meta_rev_id":"3a2a2fb110c84954c7e3474f34ff2f9b88d18f17","_number":660548,"virtual_id_number":660548,"owner":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-07-09 13:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":2,"date":"2019-07-09 13:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":1,"date":"2019-07-03 07:18:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-07-08 02:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":1,"date":"2019-07-09 13:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 12:25:00.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-05-23 12:25:03.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-06-26 01:13:10.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-07-08 02:18:18.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-07-09 13:49:58.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2019-07-09 13:58:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fc9ca26a7a1b407f0c474a266f5328110d7b2b9","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-22 02:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e175ce0a4f52357f1b2b06aeca80daeb9c7cd8e","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-22 02:51:25.000000000","message":"Topic changed from new_generic_driver to ascend_driver","accounts_in_message":[],"_revision_number":1},{"id":"9055a41a9327505eebe5c019e0b2d68c1d013f73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 03:18:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/1/check/openstack-tox-pep8/8b970de/ : FAILURE in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/1/check/openstack-tox-py27/6629c1e/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/48/660548/1/check/openstack-tox-docs/bd35981/html/ : SUCCESS in 3m 51s\n- cyborg-devstack http://logs.openstack.org/48/660548/1/check/cyborg-devstack/5f3dca6/ : SUCCESS in 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3604a0d0c98d063b100c8feb05e26fe3f227a237","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-22 10:20:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"266e0de2745be8da7154bead3acb898f910137f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:47:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/48/660548/2/check/test-release-openstack/d21dae7/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/2/check/openstack-tox-pep8/1597881/ : FAILURE in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/2/check/openstack-tox-py27/d806aee/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/48/660548/2/check/openstack-tox-docs/6b728d7/html/ : SUCCESS in 4m 07s\n- cyborg-devstack http://logs.openstack.org/48/660548/2/check/cyborg-devstack/dcd1e19/ : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54a1623953de7f149dbf4f2030d3bb94f3cf98fe","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-23 03:51:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8154d3b6e89c78f09cfb29018198b827f7730c44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 04:32:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/3/check/openstack-tox-pep8/d68b4ad/ : FAILURE in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/3/check/openstack-tox-py27/eaec667/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/48/660548/3/check/openstack-tox-docs/a48c43d/html/ : SUCCESS in 5m 24s\n- cyborg-devstack http://logs.openstack.org/48/660548/3/check/cyborg-devstack/1fde3e3/ : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"84be7b9ddc7ed7628b722ed45e92a9da6341c1ba","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-23 08:12:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d9bd63e2a616941c9a04aac1b347a12f0d0dd84","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-23 08:47:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2ceab191086a0d05160fa7beef2301fa0528c576","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:19:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/5/check/openstack-tox-pep8/dd00e1a/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/5/check/openstack-tox-py27/5cf1cd5/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/48/660548/5/check/openstack-tox-docs/d24af26/html/ : SUCCESS in 5m 28s\n- cyborg-devstack http://logs.openstack.org/48/660548/5/check/cyborg-devstack/2e1c6a8/ : SUCCESS in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c35b2d0a1b341b38d45a5f2d0fe9c7b44ca385a6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-24 16:49:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"23c6dcbbe4f046e44723fb54ca37317d157c695a","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-25 02:18:34.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"37c24f3bd8273e8b126ee8529d9bd8803e037d26","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-25 02:21:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e3445593d3899982594d7d609adf710ee4d27281","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 02:45:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/6/check/openstack-tox-pep8/4386e55/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/6/check/openstack-tox-py27/c233bb8/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/48/660548/6/check/openstack-tox-docs/a375827/html/ : SUCCESS in 4m 34s\n- cyborg-devstack http://logs.openstack.org/48/660548/6/check/cyborg-devstack/497f2f7/ : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7d1e562a6309a7edbf4799c167a246a8066002fb","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-06-18 08:08:58.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"2fc6827d04b5469bd5996d819416719c1dc60222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:45:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/7/check/openstack-tox-pep8/887eaaf/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/7/check/openstack-tox-py27/9d337fa/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/48/660548/7/check/openstack-tox-docs/1ad2e5e/html/ : SUCCESS in 4m 45s\n- cyborg-devstack http://logs.openstack.org/48/660548/7/check/cyborg-devstack/57150aa/ : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ff2012aee12e53b545b3c5956f46d1b3fc87aaa","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-26 01:13:10.000000000","message":"Patch Set 7: Code-Review+1\n\nGiving time for others to review.","accounts_in_message":[],"_revision_number":7},{"id":"af29c616d0b4edf9e47a1a5edbd474904ff8c6a5","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 07:18:45.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"affb38bc0f95d24bdc4a3060129b9cbbc87516db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:52:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/8/check/openstack-tox-pep8/cef4383/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/8/check/openstack-tox-py27/2741f97/ : SUCCESS in 2m 58s\n- openstack-tox-docs http://logs.openstack.org/48/660548/8/check/openstack-tox-docs/8c31687/html/ : SUCCESS in 2m 58s\n- cyborg-devstack http://logs.openstack.org/48/660548/8/check/cyborg-devstack/5295425/ : SUCCESS in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a6a37d21b669b59b7d8eb21f8657f9292307a281","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-08 02:18:18.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":8},{"id":"50efa1ef2a491786dfb6a2a8eab877a9a399047a","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2019-07-09 13:49:58.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"30850b890808933faa462e1e5b9878461ad61708","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 13:50:10.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"737cc2fc1b9a440dea7182a1ce6407876a50168e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 13:58:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"e1ec15b51ff7944113d50c549154c604d98759cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 13:58:34.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/660548/8/gate/openstack-tox-pep8/042ef22/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/48/660548/8/gate/openstack-tox-py27/1df6928/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/48/660548/8/gate/openstack-tox-docs/671f7d0/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e8574bb2351e2da538ce9ecfe30dfdb22e393d04","revisions":{"88094c818ffda43a98bef375d972b254aa3961f9":{"kind":"REWORK","_number":1,"created":"2019-05-22 02:47:20.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/1"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-22 02:46:53.000000000","tz":480},"subject":"WIP: Add Huawei Ascend driver","message":"WIP: Add Huawei Ascend driver\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/88094c818ffda43a98bef375d972b254aa3961f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/88094c818ffda43a98bef375d972b254aa3961f9"}]},"branch":"refs/heads/master"},"acaa13ccf6e111dbeaec1a474ac5cc1143bab08f":{"kind":"REWORK","_number":2,"created":"2019-05-22 10:20:31.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/2"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-22 10:20:03.000000000","tz":480},"subject":"WIP: Add Huawei Ascend driver","message":"WIP: Add Huawei Ascend driver\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/acaa13ccf6e111dbeaec1a474ac5cc1143bab08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/acaa13ccf6e111dbeaec1a474ac5cc1143bab08f"}]},"branch":"refs/heads/master"},"f2290791244b71167e88481fd36d7a237afa74fd":{"kind":"REWORK","_number":3,"created":"2019-05-23 03:51:15.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/3"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-23 03:49:29.000000000","tz":480},"subject":"Add NPU and GENERIC type in DB and obj","message":"Add NPU and GENERIC type in DB and obj\n\n1. Add the NPU and GENERIC type in DB model\n2. Add a migration script.\n3. Change the object to fit the new DB model\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f2290791244b71167e88481fd36d7a237afa74fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f2290791244b71167e88481fd36d7a237afa74fd"}]},"branch":"refs/heads/master"},"82e6d8c8e2c58a55891b82b2ac1a40e0eae9f48e":{"kind":"REWORK","_number":4,"created":"2019-05-23 08:12:58.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/4"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-23 07:46:01.000000000","tz":480},"subject":"Add NPU and GENERIC type in DB and obj","message":"Add NPU and GENERIC type in DB and obj\n\n1. Add the NPU and GENERIC type in DB model\n2. Add a migration script.\n3. Change the object to fit the new DB model\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/82e6d8c8e2c58a55891b82b2ac1a40e0eae9f48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/82e6d8c8e2c58a55891b82b2ac1a40e0eae9f48e"}]},"branch":"refs/heads/master"},"f563b7e32e13d32ed262a3fbbc9f519815a75132":{"kind":"REWORK","_number":5,"created":"2019-05-23 08:47:00.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/5"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-23 08:46:23.000000000","tz":480},"subject":"Add NPU and GENERIC type in DB and obj","message":"Add NPU and GENERIC type in DB and obj\n\n1. Add the NPU and GENERIC type in DB model\n2. Add a migration script.\n3. Change the object to fit the new DB model\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f563b7e32e13d32ed262a3fbbc9f519815a75132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f563b7e32e13d32ed262a3fbbc9f519815a75132"}]},"branch":"refs/heads/master"},"1a80d535756235c4a0a06aeee9947578d262b0b4":{"kind":"REWORK","_number":6,"created":"2019-05-25 02:21:00.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/6"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-25 02:17:00.000000000","tz":480},"subject":"Add NPU and GENERIC type in DB and obj","message":"Add NPU and GENERIC type in DB and obj\n\n1. Add the NPU and GENERIC type in DB model\n2. Add a migration script.\n3. Change the object to fit the new DB model\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1a80d535756235c4a0a06aeee9947578d262b0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1a80d535756235c4a0a06aeee9947578d262b0b4"}]},"branch":"refs/heads/master"},"3a5221f6a5bbb8919d171ef7709c294ec16afc60":{"kind":"REWORK","_number":7,"created":"2019-06-18 08:08:58.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/7"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-18 08:08:54.000000000","tz":0},"subject":"Add AICHIP and GENERIC type in DB and obj","message":"Add AICHIP and GENERIC type in DB and obj\n\n1. Add the AICHIP and GENERIC type in DB model\n2. Add a migration script.\n3. Change the object to fit the new DB model\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3a5221f6a5bbb8919d171ef7709c294ec16afc60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3a5221f6a5bbb8919d171ef7709c294ec16afc60"}]},"branch":"refs/heads/master"},"e8574bb2351e2da538ce9ecfe30dfdb22e393d04":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-03 07:18:45.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/48/660548/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/48/660548/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/48/660548/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/48/660548/8"}}},"commit":{"parents":[{"commit":"372591029d3e636f21499747371de085faffd5f5","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/372591029d3e636f21499747371de085faffd5f5"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-05-21 11:06:09.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-03 06:28:42.000000000","tz":480},"subject":"Add AICHIP and GENERIC type in DB and obj","message":"Add AICHIP and GENERIC type in DB and obj\n\n1. Add the AICHIP and GENERIC type in DB model\n2. Add a migration script.\n3. Change the object to fit the new DB model\n\nChange-Id: I4c7a62bba7faf081066446862b7718dee64bb7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e8574bb2351e2da538ce9ecfe30dfdb22e393d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e8574bb2351e2da538ce9ecfe30dfdb22e393d04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
