)]}'
{"id":"openstack%2Fcyborg~665318","triplet_id":"openstack%2Fcyborg~master~I08c63d533fd5dbd4330c68a22d2ecafb38134766","project":"openstack/cyborg","branch":"master","topic":"fake","hashtags":[],"change_id":"I08c63d533fd5dbd4330c68a22d2ecafb38134766","subject":"Add the fake driver","status":"MERGED","created":"2019-06-14 07:14:39.000000000","updated":"2019-08-09 22:07:46.000000000","submitted":"2019-08-09 22:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665318-1565388466143-eb0b3e6e","meta_rev_id":"b0d2bace37d9c4d9cdee50714de332a91a76db73","_number":665318,"virtual_id_number":665318,"owner":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-09 08:25:21.000000000","_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-08-09 22:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-08-09 21:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-08-09 21:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 02:14:18.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-07-27 01:24:10.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-08-09 21:57:47.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-08-09 22:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cdc9c7447c6fd42d80f9540b0beea86794b0370","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-06-14 07:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c69c6777fadf7bde267ff5005f0c596dd25d8e3","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-06-14 07:17:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"44313f95c37ec58cc37bb93fbc80fd6c3948b82d","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-06-14 07:43:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9e6d5ff9e41f111c4b63ee2bb2a8bd84f18391b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 08:19:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/665318/2/check/test-release-openstack/a122847/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/18/665318/2/check/openstack-tox-pep8/8b6e604/ : FAILURE in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/18/665318/2/check/openstack-tox-py27/faee8ba/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/18/665318/2/check/openstack-tox-docs/b418391/html/ : SUCCESS in 6m 09s\n- cyborg-devstack http://logs.openstack.org/18/665318/2/check/cyborg-devstack/02bea8e/ : SUCCESS in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da7998241d07cfdee9b6fa89f12873d5f82124f1","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-06-18 01:56:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2350222f3a9a414a5dd34982e47a57fc824310a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 02:55:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/665318/3/check/test-release-openstack/dd75a63/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/18/665318/3/check/openstack-tox-pep8/e188ad7/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/18/665318/3/check/openstack-tox-py27/9fa464d/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/18/665318/3/check/openstack-tox-docs/474d0ae/html/ : SUCCESS in 5m 05s\n- cyborg-devstack http://logs.openstack.org/18/665318/3/check/cyborg-devstack/12c3b9b/ : SUCCESS in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eae5fe213f095d6aa04180f3d587ca6a87638ee0","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-26 01:14:19.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a60a98f916851667ab61dd76782292fa2ef180a3","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-01 03:19:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"232e645ba161c8db46d0fceea3ad7df197ab9c89","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-03 03:27:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"400a50d855b14bc8f98fa59353964a4b60b0414b","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 06:22:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5aa88755b1e88b2c82d0d31941b79cde4ceb3188","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 06:22:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3776b24e1db24f9d9b6271db4922ed8c9735dafb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:41:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/665318/4/check/test-release-openstack/11cf7f8/ : SUCCESS in 1m 41s\n- openstack-tox-pep8 http://logs.openstack.org/18/665318/4/check/openstack-tox-pep8/e2c3a41/ : FAILURE in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/18/665318/4/check/openstack-tox-py27/10a63b9/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/18/665318/4/check/openstack-tox-docs/53f7c19/html/ : SUCCESS in 4m 13s\n- cyborg-devstack http://logs.openstack.org/18/665318/4/check/cyborg-devstack/c02fe1f/ : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b87c25eb178478dae5f1e1820df486a0fcc7829","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 08:06:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1ad370cccd650e706c8de266c5c86536467bb142","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 08:06:30.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3adb8d482e97ec90a464c3fc0506092546087b8e","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 08:07:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"e5ee96e711a6185cf19fd5e4288a4040a0a36d0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 08:45:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/665318/7/check/test-release-openstack/441f2bc/ : SUCCESS in 2m 14s\n- openstack-tox-pep8 http://logs.openstack.org/18/665318/7/check/openstack-tox-pep8/0647b87/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/18/665318/7/check/openstack-tox-py27/c297e05/ : SUCCESS in 3m 03s\n- openstack-tox-docs http://logs.openstack.org/18/665318/7/check/openstack-tox-docs/1fa66f1/html/ : SUCCESS in 3m 00s\n- cyborg-devstack http://logs.openstack.org/18/665318/7/check/cyborg-devstack/272cc51/ : SUCCESS in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"278f4eccdba63666d9bce0b7a79d89c43175a1a6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-04 13:24:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d7da9c6bf120f5e587c1f2cad11120ed749b1251","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-08 01:44:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cd67b2241df422d3ff82ea8d5b4f4040d81ea02c","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-08 02:14:18.000000000","message":"Patch Set 7: Code-Review-1\n\nAlmost looks good. \nOne thing need to be pointed out: driver should fill attribute_list with resource class and traits if needed. This allow conductor to extract them and report them to placement. Here is an example:\nhttps://github.com/openstack/cyborg/blob/master/cyborg/accelerator/drivers/fpga/intel/sysinfo.py#L246","accounts_in_message":[],"_revision_number":7},{"id":"70e3f37dc76083d2185a83ee88d99c7d5ad0cb31","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-10 07:25:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"07cd50f85a52b615562d4d5e28e7adef9812c157","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-10 07:42:00.000000000","message":"Patch Set 8:\n\n\u003e Almost looks good.\n \u003e One thing need to be pointed out: driver should fill attribute_list\n \u003e with resource class and traits if needed. This allow conductor to\n \u003e extract them and report them to placement. Here is an example:\n \u003e https://github.com/openstack/cyborg/blob/master/cyborg/accelerator/drivers/fpga/intel/sysinfo.py#L246\n\ndone, thanks for your comments","accounts_in_message":[],"_revision_number":8},{"id":"0060dcef8cf9b235764c7051b1827ae197844754","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 07:57:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/665318/8/check/test-release-openstack/69ee240/ : SUCCESS in 2m 11s\n- openstack-tox-pep8 http://logs.openstack.org/18/665318/8/check/openstack-tox-pep8/490078e/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/18/665318/8/check/openstack-tox-py27/e86f009/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/18/665318/8/check/openstack-tox-docs/8236656/html/ : SUCCESS in 3m 12s\n- cyborg-devstack http://logs.openstack.org/18/665318/8/check/cyborg-devstack/9058043/ : SUCCESS in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6bb21021394b2dc4fef63ab4c1458c8730a5567a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-11 22:44:23.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nThanks, yikun.","accounts_in_message":[],"_revision_number":8},{"id":"edf8f82ccefa524482da0e80b81af36d09eec0f1","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-25 02:10:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"05f871c4df57ace7a4675428d2bf9d0c0d7d86c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 02:42:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/665318/9/check/test-release-openstack/87b30af/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/18/665318/9/check/openstack-tox-pep8/3efefe4/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/18/665318/9/check/openstack-tox-py27/1c408f9/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/18/665318/9/check/openstack-tox-docs/cbdfec7/html/ : SUCCESS in 4m 03s\n- cyborg-devstack http://logs.openstack.org/18/665318/9/check/cyborg-devstack/14042a6/ : SUCCESS in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d36a4fa65895a7e3cf37e2b0fbe54766f2c028ef","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-27 20:30:29.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\nThis is getting close to the end, I think.","accounts_in_message":[],"_revision_number":9},{"id":"65f094145b77cf4361b7e97f89b736858dcf374f","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-08-09 07:48:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"017ed12355f2a428eef5fc37d8d5c6ed6f965e7f","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 08:07:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"597b5375fa9c08ff0df75eb0f2a114626bdf8af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 08:23:01.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/18/665318/10/check/test-release-openstack/ac917c3/ : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://logs.opendev.org/18/665318/10/check/openstack-tox-pep8/ba0326f/ : FAILURE in 2m 46s\n- openstack-tox-py27 https://logs.opendev.org/18/665318/10/check/openstack-tox-py27/2d52331/ : SUCCESS in 4m 51s\n- requirements-check https://logs.opendev.org/18/665318/10/check/requirements-check/e2ef4a5/ : SUCCESS in 3m 14s\n- openstack-tox-docs https://logs.opendev.org/18/665318/10/check/openstack-tox-docs/0b9f784/html/ : SUCCESS in 4m 29s\n- cyborg-devstack https://logs.opendev.org/18/665318/10/check/cyborg-devstack/0d39fe1/ : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9174035966867b8bb2c77909f95966fbfa9a9dd6","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-08-09 08:25:11.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"cef7a102bbb0a81ea34fa287fdca5ce13adcde29","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-08-09 08:25:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a3a946b105a2a54aa65261b8f798319ef495c637","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 08:57:54.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/18/665318/11/check/test-release-openstack/1a35474/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://logs.opendev.org/18/665318/11/check/openstack-tox-pep8/6420655/ : SUCCESS in 3m 59s\n- openstack-tox-py27 https://logs.opendev.org/18/665318/11/check/openstack-tox-py27/5df6ff8/ : SUCCESS in 3m 45s\n- requirements-check https://logs.opendev.org/18/665318/11/check/requirements-check/9841204/ : SUCCESS in 1m 58s\n- openstack-tox-docs https://logs.opendev.org/18/665318/11/check/openstack-tox-docs/90e8348/html/ : SUCCESS in 3m 43s\n- cyborg-devstack https://logs.opendev.org/18/665318/11/check/cyborg-devstack/2b1ecc8/ : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"332f13ba1130437ba4c0b1de504ecefc9ad0fad1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 21:57:41.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7f359ec01b665ed0e3e56179b403c283568528d6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 21:57:47.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"87fd44ba8e3d6b45f29b87ea0b95574475ff3031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 21:57:59.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"55ba6f928e36a5b13da5e74d660e2b39182c43b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 22:07:45.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://logs.opendev.org/18/665318/11/gate/test-release-openstack/027caac/ : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://logs.opendev.org/18/665318/11/gate/openstack-tox-pep8/3de3949/ : SUCCESS in 3m 10s\n- openstack-tox-py27 https://logs.opendev.org/18/665318/11/gate/openstack-tox-py27/92ee940/ : SUCCESS in 3m 10s\n- requirements-check https://logs.opendev.org/18/665318/11/gate/requirements-check/ff8badb/ : SUCCESS in 2m 19s\n- openstack-tox-docs https://logs.opendev.org/18/665318/11/gate/openstack-tox-docs/f0d11a4/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":11},{"id":"f11699ab1711924f6e8dd1bfef5ef0aadd8710fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 22:07:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"658334455bbd113611097645f2b9c2f89ae98527","revisions":{"de3bdc4f5c13dbcbf8163682bb8a34e18656e818":{"kind":"REWORK","_number":1,"created":"2019-06-14 07:14:39.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/1"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/de3bdc4f5c13dbcbf8163682bb8a34e18656e818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/de3bdc4f5c13dbcbf8163682bb8a34e18656e818"}]},"branch":"refs/heads/master"},"626f096e2e1499e2dfae3c71cef9710b4a92712f":{"kind":"REWORK","_number":2,"created":"2019-06-14 07:43:50.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/2"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:43:21.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/626f096e2e1499e2dfae3c71cef9710b4a92712f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/626f096e2e1499e2dfae3c71cef9710b4a92712f"}]},"branch":"refs/heads/master"},"081b7bb40ec275ec3ed3c4a376685ad7a08136e2":{"kind":"REWORK","_number":3,"created":"2019-06-18 01:56:17.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/3"}}},"commit":{"parents":[{"commit":"99bb44399f1d0a91d10c51a03ce63708f0651c50","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/99bb44399f1d0a91d10c51a03ce63708f0651c50"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-18 01:55:10.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/081b7bb40ec275ec3ed3c4a376685ad7a08136e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/081b7bb40ec275ec3ed3c4a376685ad7a08136e2"}]},"branch":"refs/heads/master"},"49d030778f743e924df5eb7b905cafa58ec8ddc4":{"kind":"REWORK","_number":4,"created":"2019-07-03 06:22:41.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/4"}}},"commit":{"parents":[{"commit":"22fdd66f770b77f10fb098d32e0d6c706ad4e674","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/22fdd66f770b77f10fb098d32e0d6c706ad4e674"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-03 06:22:15.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/49d030778f743e924df5eb7b905cafa58ec8ddc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/49d030778f743e924df5eb7b905cafa58ec8ddc4"}]},"branch":"refs/heads/master"},"3323f55d5d04f43bffa48e9f430df8d9e2163c14":{"kind":"REWORK","_number":5,"created":"2019-07-03 08:06:21.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/5"}}},"commit":{"parents":[{"commit":"22fdd66f770b77f10fb098d32e0d6c706ad4e674","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/22fdd66f770b77f10fb098d32e0d6c706ad4e674"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-03 08:05:48.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3323f55d5d04f43bffa48e9f430df8d9e2163c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3323f55d5d04f43bffa48e9f430df8d9e2163c14"}]},"branch":"refs/heads/master"},"9bb10b84313bc015cc2587eccf936f3dfa84cb8a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-03 08:06:30.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/6"}}},"commit":{"parents":[{"commit":"75ecc3ab85f49faad74bb8d3be1f0a9aa6d752c6","subject":"Update json module to jsonutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/75ecc3ab85f49faad74bb8d3be1f0a9aa6d752c6"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-03 08:06:30.000000000","tz":0},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9bb10b84313bc015cc2587eccf936f3dfa84cb8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9bb10b84313bc015cc2587eccf936f3dfa84cb8a"}]},"branch":"refs/heads/master"},"15643de70f4906111b0a4af767c2a6c0dcd87e8c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-03 08:07:21.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/7"}}},"commit":{"parents":[{"commit":"372591029d3e636f21499747371de085faffd5f5","subject":"Add the new generic driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/372591029d3e636f21499747371de085faffd5f5"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-03 08:07:21.000000000","tz":0},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/15643de70f4906111b0a4af767c2a6c0dcd87e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/15643de70f4906111b0a4af767c2a6c0dcd87e8c"}]},"branch":"refs/heads/master"},"6553c510bfabaaf74a3cd191843342e36ce82738":{"kind":"REWORK","_number":8,"created":"2019-07-10 07:25:34.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/8"}}},"commit":{"parents":[{"commit":"74c68900bac8f1413ecc608af011686eaee3661d","subject":"Merge \"Change attach_info, cpid_info from string to a dict.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/74c68900bac8f1413ecc608af011686eaee3661d"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-10 07:24:34.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6553c510bfabaaf74a3cd191843342e36ce82738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6553c510bfabaaf74a3cd191843342e36ce82738"}]},"branch":"refs/heads/master"},"30963cd840c7332b10963ba81c9760cf3f554236":{"kind":"REWORK","_number":9,"created":"2019-07-25 02:10:08.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/9"}}},"commit":{"parents":[{"commit":"74c68900bac8f1413ecc608af011686eaee3661d","subject":"Merge \"Change attach_info, cpid_info from string to a dict.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/74c68900bac8f1413ecc608af011686eaee3661d"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-25 02:09:39.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/30963cd840c7332b10963ba81c9760cf3f554236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/30963cd840c7332b10963ba81c9760cf3f554236"}]},"branch":"refs/heads/master"},"65607fe9aafa1053a8b496b8148631ba83021bd1":{"kind":"REWORK","_number":10,"created":"2019-08-09 07:48:49.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/10"}}},"commit":{"parents":[{"commit":"74c68900bac8f1413ecc608af011686eaee3661d","subject":"Merge \"Change attach_info, cpid_info from string to a dict.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/74c68900bac8f1413ecc608af011686eaee3661d"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-08-09 07:48:21.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/65607fe9aafa1053a8b496b8148631ba83021bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/65607fe9aafa1053a8b496b8148631ba83021bd1"}]},"branch":"refs/heads/master"},"658334455bbd113611097645f2b9c2f89ae98527":{"kind":"REWORK","_number":11,"created":"2019-08-09 08:25:11.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/18/665318/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/665318/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/665318/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/665318/11"}}},"commit":{"parents":[{"commit":"74c68900bac8f1413ecc608af011686eaee3661d","subject":"Merge \"Change attach_info, cpid_info from string to a dict.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/74c68900bac8f1413ecc608af011686eaee3661d"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-06-14 07:13:57.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-08-09 08:24:53.000000000","tz":480},"subject":"Add the fake driver","message":"Add the fake driver\n\nThis patch add the fake driver, this driver can be configured\nin cyborg.cfg to enable it:\n\n[agent]\nenabled_drivers \u003d fake_driver\n\nIt would be used in cyborg test.\n\nChange-Id: I08c63d533fd5dbd4330c68a22d2ecafb38134766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/658334455bbd113611097645f2b9c2f89ae98527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/658334455bbd113611097645f2b9c2f89ae98527"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
