)]}'
{"id":"openstack%2Fcyborg~667524","triplet_id":"openstack%2Fcyborg~master~Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12","project":"openstack/cyborg","branch":"master","topic":"python3","hashtags":[],"change_id":"Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12","subject":"python3 patch (including zuul config)","status":"MERGED","created":"2019-06-26 06:50:57.000000000","updated":"2019-09-13 16:08:07.000000000","submitted":"2019-09-13 16:03:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667524-1568390635322-7da7b55f","meta_rev_id":"86ae029ab26d4eae4bbe8be9f9021a554f7c6f23","_number":667524,"virtual_id_number":667524,"owner":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-09-13 16:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"recommended":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":1,"date":"2019-09-13 14:30:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-09-13 14:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2019-09-13 15:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-09-13 14:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-13 14:30:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-09-13 14:30:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":1,"date":"2019-09-13 15:16:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 18:02:33.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-08-26 01:40:17.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2019-08-26 01:41:18.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-09-05 11:24:17.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-09-10 02:00:50.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-09-12 07:17:50.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2019-09-13 03:42:12.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-09-13 05:10:20.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-09-13 15:07:16.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-13 15:16:51.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-09-13 16:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89f45725aafdf00db6c810871f6fcb071dafb798","author":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"date":"2019-06-26 06:50:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b0c0464c4e9dca97c95f195f4db89a5fc5cde72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 07:47:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/24/667524/1/check/openstack-tox-pep8/d90fc1d/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/24/667524/1/check/openstack-tox-py27/c01540e/ : SUCCESS in 3m 01s\n- openstack-tox-docs http://logs.openstack.org/24/667524/1/check/openstack-tox-docs/750690d/html/ : SUCCESS in 3m 54s\n- cyborg-devstack http://logs.openstack.org/24/667524/1/check/cyborg-devstack/08b781b/ : SUCCESS in 29m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da93a0fb340534f52f8d51682bbefdac91152ab9","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-07-29 08:46:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc2025994854de048d626e88caf4c5756c6edf83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 09:24:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/24/667524/2/check/openstack-tox-pep8/d98ccda/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/24/667524/2/check/openstack-tox-py27/ddb0b80/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/24/667524/2/check/openstack-tox-py36/62a5d83/ : FAILURE in 3m 35s (non-voting)\n- openstack-tox-py37 http://logs.openstack.org/24/667524/2/check/openstack-tox-py37/85bfe70/ : FAILURE in 3m 48s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/24/667524/2/check/openstack-tox-docs/9ff56d4/html/ : SUCCESS in 3m 38s\n- cyborg-devstack http://logs.openstack.org/24/667524/2/check/cyborg-devstack/b512fd7/ : SUCCESS in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36fde5ebc112d05601d227e6f1c01a63b7dd4b94","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-05 06:29:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d15eb22cd69ffacddc277e110d9a0297de8e817","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 07:06:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/24/667524/3/check/openstack-tox-pep8/bc932a5/ : FAILURE in 4m 29s\n- openstack-tox-py27 https://logs.opendev.org/24/667524/3/check/openstack-tox-py27/9cf3a79/ : SUCCESS in 4m 57s\n- openstack-tox-docs https://logs.opendev.org/24/667524/3/check/openstack-tox-docs/197578c/html/ : SUCCESS in 3m 38s\n- cyborg-devstack https://logs.opendev.org/24/667524/3/check/cyborg-devstack/e6d249d/ : SUCCESS in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a386739ecf4501f23c93f09d5edd8ba99bfabce","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-05 07:38:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c56625a958525bc46cfb43cfef6f879adb50f9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 08:17:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/24/667524/4/check/openstack-tox-pep8/ca93919/ : FAILURE in 4m 39s\n- openstack-tox-py27 https://logs.opendev.org/24/667524/4/check/openstack-tox-py27/7e18c79/ : SUCCESS in 6m 24s\n- openstack-tox-py36 https://logs.opendev.org/24/667524/4/check/openstack-tox-py36/ed95537/ : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py37 https://logs.opendev.org/24/667524/4/check/openstack-tox-py37/8f4cb51/ : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/24/667524/4/check/openstack-tox-docs/aa8eb7a/html/ : SUCCESS in 3m 43s\n- cyborg-devstack https://logs.opendev.org/24/667524/4/check/cyborg-devstack/fbd0a52/ : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"85e0492888075a61473801a356a8cc517a2dccce","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-05 08:45:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"91433cb264e1ebef0c0f0fa65a43dc88884e667d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:24:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/24/667524/5/check/openstack-tox-pep8/034ab3b/ : SUCCESS in 4m 48s\n- openstack-tox-py27 https://logs.opendev.org/24/667524/5/check/openstack-tox-py27/662a855/ : SUCCESS in 5m 20s\n- openstack-tox-py36 https://logs.opendev.org/24/667524/5/check/openstack-tox-py36/d40c36e/ : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py37 https://logs.opendev.org/24/667524/5/check/openstack-tox-py37/0e279bf/ : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/24/667524/5/check/openstack-tox-docs/13d8aae/html/ : SUCCESS in 4m 29s\n- cyborg-devstack https://logs.opendev.org/24/667524/5/check/cyborg-devstack/81c2e38/ : SUCCESS in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f1c2ccd62cba7f9c756f81754ca52d230aa956c1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-05 20:57:27.000000000","message":"Patch Set 5: Code-Review+1\n\nGiving time for others to review.","accounts_in_message":[],"_revision_number":5},{"id":"e71618cb6dba8ee880cd688c2eab9ea8a4e91ab0","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-12 18:02:33.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6551fa6cf1b858ac9081ec5012ce122ef5fbfafd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-12 22:41:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a8b09a768553f573d5fc6a1e84079bba3841cbd0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-19 01:48:24.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nlet\u0027s go with voting job otherwise it will be more changes in future if you make it n-v and allow the py3 breaking changes in.","accounts_in_message":[],"_revision_number":5},{"id":"216979c11e212afa7acee4287fd5a3ad62d2590a","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-22 08:12:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9bca98d693ce1c43f349d63a4c7571e9eca1f939","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:48:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352620e0ea3b433fbaf8450ff6342aaf : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/020140f2681442378d1083e5d692d38a : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d3548b12bb47ae8a4c0ff8c55d6af7 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6fc5a7c3bec4273a14cab6cd7e1b13e : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a67971a29d4d0b8c4f1fd5803bc76e : FAILURE in 4m 18s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/5e73919689b748909d1e6319370f7841 : SUCCESS in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e056faf3e30f007d9f39a69b12eb33d6bd365119","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-22 10:36:45.000000000","message":"Patch Set 6:\n\n\u003e let\u0027s go with voting job otherwise it will be more changes in\n \u003e future if you make it n-v and allow the py3 breaking changes in.\n\nWe have a 2-phase plan where this patch addresses some things and we cover the rest with a subsequent patch. Cyborg community has discussed and decided to make this a non-voting job for now. I understand your concern, though. We _should_ make this a voting job after merging the important changes by Train-3 milestone and then proceed to fix the bugs.","accounts_in_message":[],"_revision_number":6},{"id":"bffcc319834cc41e29621f69f9dbecfb60a2146b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-22 10:38:28.000000000","message":"Patch Set 6: Code-Review-1\n\nThanks, Shogo. Please add the Py3.7 support in setup.cfg as per [1].\n\n[1] https://review.opendev.org/#/c/675990/1/","accounts_in_message":[],"_revision_number":6},{"id":"4b5ae16e436d14acf9f30ff5ed14b464eb19e954","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-22 11:03:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3a31821264a0480afb304c8139738f2ddd326f17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:35:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa7cd58a92c4451ebb74b747b00597d8 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cba97a384bc4e39b46698b465766a00 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7c5b4992ecc45ac977e21946b242e4b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24a2b1972ba644ccb77f9f9446c26e30 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5426ddcf2f8a4ee98b5b2ceac704c238 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f9adab9b93447aba819a2e8e6ab9054 : FAILURE in 4m 11s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/1b63bb61943d415588bb9f045b3174f7 : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"11139cd3a25ff2c1b4c73bbcaafe4fa9a8fc6ef3","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-23 01:03:22.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"7bb63aa00771a6dabe2be5dbd18d0170771b7d98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:31:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2695016473b144a6affcecb21ce86476 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e8ec78eb229496cb36123be208900c1 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3165427a23134a25902cd830e5b5d598 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae13c54b22f463da37b962fa3b31dc7 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46fdc72a528a425dab29ee0dd337fab2 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af51f5d0959743ab90aa259d40bb95de : SUCCESS in 3m 52s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/688647885b0640188c48ea0fd3a95a91 : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"70f9a80574a42ec7c75036a8943b0d8bd576a739","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-26 01:27:04.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c93b9683eaff5f02d98acdd883126255bfdb6d21","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-26 01:40:17.000000000","message":"Patch Set 8:\n\n\u003e (1 comment)\n\nAs Sundar mentioned above, this config is temporary one. We should modify them to openstack-python3-train-jobs in following patch, by the Train-3 release.","accounts_in_message":[],"_revision_number":8},{"id":"757263d69119e6244e60b979b87d542a978c4a72","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-26 01:41:18.000000000","message":"Patch Set 8: -Code-Review\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e As Sundar mentioned above, this config is temporary one. We should\n \u003e modify them to openstack-python3-train-jobs in following patch, by\n \u003e the Train-3 release.\n\nOK, anyway","accounts_in_message":[],"_revision_number":8},{"id":"22ea56e94043d90010bc31925b3a89bae34f97d5","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-27 09:15:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8c74648c2048696eb594e51a1513e1c68e0daadf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 09:55:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/969db66c1d8b4590bbba0f47163632f9 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e30df05eea4ad9a4a944388635a422 : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10a6779e3bef48ec900caac42b141359 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3100110b8644fe97b23d3abebf4057 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/100ab7b9507242ffb328ff1ae551d910 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1998c5bf95c34bf8b1e600873df45ac0 : SUCCESS in 6m 06s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/f933136adffe4bb08a6f812a90ad184e : SUCCESS in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8297a9ea8607346d7ecced5ae5187a9b097d37bc","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-27 17:04:42.000000000","message":"Patch Set 9: Code-Review+1\n\nLooks good overall. Waiting for other reviews and also for P5-P9 patches to merge, before giving +2.","accounts_in_message":[],"_revision_number":9},{"id":"b45ee8b4f808e43fe9f0f79171e1425304d8a3fd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 00:43:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bcdade2f132b810fca830b9be23d992e244734e4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 01:09:44.000000000","message":"Topic changed from nova-integ to python3","accounts_in_message":[],"_revision_number":10},{"id":"84fd728d7fefcb847789801ed72558ffeeaa1028","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 01:16:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/67f74c65eb754a6aa7500b91332ac21e : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f26d8fb3f3e040b5966f33ab91d49915 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/983fd5a170f64f788cda64e8572a7f33 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/357737ce39074bc5953e61177a561a85 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7485ab3140945f097c97c0285bc1931 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83222832aaf4a608da3a64df6fc5c93 : SUCCESS in 4m 08s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/465d2ddb9fe049d08edfe82312346d57 : SUCCESS in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"470535400856454dff0e7e1276a10ed39f57834d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-10 02:00:50.000000000","message":"Patch Set 10: Code-Review+1\n\nThanks. lgtm","accounts_in_message":[],"_revision_number":10},{"id":"dffbf7acd3435493d5ce22878947d8143e6b5ac0","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-09-12 07:17:50.000000000","message":"Patch Set 10: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":10},{"id":"de75abcd62f9a61006ef6a03a9373bc3ede3aa9d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-12 18:51:29.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"b464f5e97cece6888e0db4b39494aad98a79b363","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-12 19:04:53.000000000","message":"Patch Set 10: -Workflow","accounts_in_message":[],"_revision_number":10},{"id":"7a7d95fcaa16f3859636166f8d2d548297a295a5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-13 03:18:52.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"534398d7dcc9536d604d72fd010b39b674fca614","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-09-13 03:42:12.000000000","message":"Patch Set 10: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"cd724f94bebcb4a5dfd7293dfc6d4297f0030062","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-09-13 05:10:20.000000000","message":"Patch Set 10: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"4ddb13191cffa3db5a16f8cf43864ee9780ffe96","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-09-13 05:18:40.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"aaad5a190e0c81054314c7899574fe9cadabd097","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-13 09:29:46.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a46ee4978fdab321ceb8c4cb9ed4b56a1f8c222b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-13 14:30:51.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"5416f0377e6568b3b8fe6e7413c2777462ef1613","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 15:07:16.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"dccca4c91e8f478077ce88777ffe60cd9a6edb7d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-13 15:16:51.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"e34cdfed8a0ba07dbcecfb023e6394068d07ac6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:22:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b6a4d1f8f5149398593a405bc06b38c : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3443b52ad1804d1fa5a267d0a2e2b367 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48e483eb8d854b58a4148b5c0fd7fe66 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/274046923bf84ec2bbb9ea44a12e2427 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a968faf1c9e411cb1f0ee0e5b4953c5 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b52eb230dde466d96814711b3545877 : SUCCESS in 3m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4cb4d24decd14884a076d79769300442 : SUCCESS in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"05dbdf1f8d1b6e3ae7ecf1ec0c40d70c7baad58d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:23:09.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"886578078c1f1118ffe4bf8db654ca1d93ba2f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:03:55.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/59955f7a3efb4f62ba957779ea737fe8 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7905e2dffe455a8650731f56897450 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf8badea8c60433bac9efbc7a9c33b92 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7161e98899a144daa4b5f93451728d73 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1db4c1e4042842b9ba37ccdbcc230be5 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e33c3f2189f4f3891e85632a03fec05 : SUCCESS in 5m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6414ea09da684e1c9d0e7623147c1ca5 : SUCCESS in 35m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1a003e9d4f9a9dd8a2d9551539f0ca348a2a3a38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:03:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"86ae029ab26d4eae4bbe8be9f9021a554f7c6f23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:08:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9d089e5a294e01b1c9bb4cbf543360 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"38119f675ea9e99a0045c63490243f1561c4cb9c","revisions":{"d5b958d3639a78b3894a19232982bfc284f0814f":{"kind":"REWORK","_number":1,"created":"2019-06-26 06:50:57.000000000","uploader":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"ref":"refs/changes/24/667524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/1"}}},"commit":{"parents":[{"commit":"553ca3bb0352de1cd6d7618277ad1f316504fdf7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/553ca3bb0352de1cd6d7618277ad1f316504fdf7"}]}],"author":{"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","date":"2019-06-26 06:51:56.000000000","tz":540},"committer":{"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","date":"2019-06-26 06:52:21.000000000","tz":540},"subject":"Python3 migration 1st patch","message":"Python3 migration 1st patch\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d5b958d3639a78b3894a19232982bfc284f0814f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d5b958d3639a78b3894a19232982bfc284f0814f"}]},"branch":"refs/heads/master"},"eb4a5d7795c9662a5561aed4500a7c26cea7eb91":{"kind":"REWORK","_number":2,"created":"2019-07-29 08:46:54.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/2"}}},"commit":{"parents":[{"commit":"16c6c378a96825049b786a60678015c5fedf01c9","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/16c6c378a96825049b786a60678015c5fedf01c9"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-07-29 08:35:50.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-07-29 08:43:28.000000000","tz":540},"subject":"WIP: python3 patch with zuul test. python2/3 patch.","message":"WIP: python3 patch with zuul test.\npython2/3 patch.\n\npython2/3 migration patch.\n(it still keep compatibility with python2.)\nfixed some points and zuul config.\n\nFix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib. (cyborg/common/exception.py)\n\n2nd: check \u0026 fix manually.\n2-1:\nas_dict() function ( cyborg/objects/base.py)\nfix as_dict() while reffering other project.\n\n2-2:\ndelete unnecessary as_dict() call in response to the change in as_dict()\n(cyborg/tests/unit/accelerator/drivers/fpga/intel/test_driver.py)\n(cyborg/tests/unit/accelerator/drivers/gpu/test_utils.py)\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eb4a5d7795c9662a5561aed4500a7c26cea7eb91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eb4a5d7795c9662a5561aed4500a7c26cea7eb91"}]},"branch":"refs/heads/master"},"b68d29183599f4f6d22368ece7911418e56d8173":{"kind":"REWORK","_number":3,"created":"2019-08-05 06:29:54.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/3"}}},"commit":{"parents":[{"commit":"0d99c026c475339127df70da38e3db418058a949","subject":"Merge \"Mailing lists change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0d99c026c475339127df70da38e3db418058a949"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-05 06:19:44.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-05 06:22:25.000000000","tz":540},"subject":"WIP: python3 patch with zuul test","message":"WIP: python3 patch with zuul test\n\nThis patch doesn\u0027t need review and merge,\nbecause this patch will be tested and posted again,\nafter P1-P8 patches are merged.\n\npython2/3 migration patch ( still keep compatibility with python2.)\n\nfixed some points and zuul config.\n\nFix points are extracted in following methods.\n\n1st: check with sixer:\n\tchange unicode function to use sixer lib.\n\t(cyborg/common/exception.py)\n\n2nd: check \u0026 fix manually.\n\n2-1: as_dict() function ( cyborg/objects/base.py)\n\tfix as_dict() while reffering other project.\n\n2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n\t(cyborg/tests/unit/accelerator/drivers/fpga/intel/test_driver.py)\n\t(cyborg/tests/unit/accelerator/drivers/gpu/test_utils.py)\n\n2-3: fix all_pfs_have_vf()  to be able to treat both  list and iterator,\n\tbecause filter() function returns list in py2, iterator in py3.\n\n\t(/cyborg/accelerator/drivers/fpga/intel/sysinfo.py)\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b68d29183599f4f6d22368ece7911418e56d8173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b68d29183599f4f6d22368ece7911418e56d8173"}]},"branch":"refs/heads/master"},"a10b626a2772302ba92b32d731d88ad8c9daa3f4":{"kind":"REWORK","_number":4,"created":"2019-08-05 07:38:54.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/4"}}},"commit":{"parents":[{"commit":"0d99c026c475339127df70da38e3db418058a949","subject":"Merge \"Mailing lists change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0d99c026c475339127df70da38e3db418058a949"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-05 07:35:09.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-05 07:35:50.000000000","tz":540},"subject":"WIP: python3 patch with zuul test","message":"WIP: python3 patch with zuul test\n\nThis patch doesn\u0027t need review and merge,\nbecause this patch will be tested and posted again,\nafter P1-P8 patches are merged.\n\npython2/3 migration patch ( still keep compatibility with python2.)\n\nfixed some points and zuul config.\n\nFix points are extracted in following methods.\n\n1st: check with sixer:\n\tchange unicode function to use sixer lib.\n\t(cyborg/common/exception.py)\n\n2nd: check \u0026 fix manually.\n\n2-1: as_dict() function ( cyborg/objects/base.py)\n\tfix as_dict() while reffering other project.\n\n2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n\t(cyborg/tests/unit/accelerator/drivers/fpga/intel/test_driver.py)\n\t(cyborg/tests/unit/accelerator/drivers/gpu/test_utils.py)\n\n2-3: fix all_pfs_have_vf()  to be able to treat both  list and iterator,\n\tbecause filter() function returns list in py2, iterator in py3.\n\n\t(/cyborg/accelerator/drivers/fpga/intel/sysinfo.py)\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a10b626a2772302ba92b32d731d88ad8c9daa3f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a10b626a2772302ba92b32d731d88ad8c9daa3f4"}]},"branch":"refs/heads/master"},"0e041fd42257e87d5dfd4147b3c538688de50c66":{"kind":"REWORK","_number":5,"created":"2019-08-05 08:45:15.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/5"}}},"commit":{"parents":[{"commit":"0d99c026c475339127df70da38e3db418058a949","subject":"Merge \"Mailing lists change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0d99c026c475339127df70da38e3db418058a949"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-05 08:40:13.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-05 08:40:52.000000000","tz":540},"subject":"WIP: python3 patch with zuul test python2/3 migration patch ( still keep compatibility with python2)","message":"WIP: python3 patch with zuul test\npython2/3 migration patch ( still keep compatibility with python2)\n\n** This patch doesn\u0027t need review and merge **\nthis patch will be tested and posted again,\nafter P1-P8 patches are merged.\n\nfixed some points and zuul config.\n\nFix points are extracted in following methods.\n\n1st: check with sixer:\n\tchange unicode function to use sixer lib.\n\t(cyborg/common/exception.py)\n\n2nd: check \u0026 fix manually.\n\n2-1: as_dict() function ( cyborg/objects/base.py)\n\tfix as_dict() while reffering other project.\n\n2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n\t(cyborg/tests/unit/accelerator/drivers/fpga/intel/test_driver.py)\n\t(cyborg/tests/unit/accelerator/drivers/gpu/test_utils.py)\n\n2-3: fix all_pfs_have_vf()  to be able to treat both  list and iterator,\n\tbecause filter() function returns list in py2, iterator in py3.\n\t(/cyborg/accelerator/drivers/fpga/intel/sysinfo.py)\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0e041fd42257e87d5dfd4147b3c538688de50c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0e041fd42257e87d5dfd4147b3c538688de50c66"}]},"branch":"refs/heads/master"},"c366e581f5262d3f0c393e19d0484cc66f0d22d6":{"kind":"REWORK","_number":6,"created":"2019-08-22 08:12:39.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/6"}}},"commit":{"parents":[{"commit":"40dad1b19f4f07fb04d66b4a3743e6766f7658ad","subject":"P9: Enable v2 API in devstack.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40dad1b19f4f07fb04d66b4a3743e6766f7658ad"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"subject":"python3 patch (including zuul config)","message":"python3 patch (including zuul config)\n\n Fix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib. (cyborg/common/exception.py)\n2nd: check \u0026 fix manually.\n    2-1: as_dict() function ( cyborg/objects/base.py)\n         fix as_dict() while reffering other project.\n    2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n         (/tests/unit/accelerator/drivers/fpga/intel/test_driver.py)\n        (/tests/unit/accelerator/drivers/gpu/test_utils.py)\n    2-3: fix all_pfs_have_vf() to be able to treat both list and iterator,\n         because filter() function returns list in py2, iterator in py3.\n        (/accelerator/drivers/fpga/intel/sysinfo.py)\n    2-4: wrap map() with list()  in P6,P7\n        (/tests/unit/fake_device_profile.py )\n        (/tests/unit/fake_extarq.py )\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\nStory: #2003219\nTask: #23418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c366e581f5262d3f0c393e19d0484cc66f0d22d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c366e581f5262d3f0c393e19d0484cc66f0d22d6"}]},"branch":"refs/heads/master"},"772c9742099648966eaaa72b879c4afaf3c8b89f":{"kind":"REWORK","_number":7,"created":"2019-08-22 11:03:31.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/7"}}},"commit":{"parents":[{"commit":"40dad1b19f4f07fb04d66b4a3743e6766f7658ad","subject":"P9: Enable v2 API in devstack.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40dad1b19f4f07fb04d66b4a3743e6766f7658ad"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 10:57:36.000000000","tz":540},"subject":"python3 patch (including zuul config)","message":"python3 patch (including zuul config)\n\n Fix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib.\n2nd: check \u0026 fix manually.\n    2-1: as_dict() function\n         fix as_dict() to work in py3 env.\n    2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n    2-3: fix all_pfs_have_vf() to be able to treat both list and iterator,\n         because filter() function returns list in py2, iterator in py3.\n    2-4: To treats map() in same way in py2/3 env , wrap map() with list()  in P6,P7\n    2-5: add python3.7 to setup.cfg\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\nStory: #2003219\nTask: #23418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/772c9742099648966eaaa72b879c4afaf3c8b89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/772c9742099648966eaaa72b879c4afaf3c8b89f"}]},"branch":"refs/heads/master"},"0430c4bd7190dd161138e688bfd1293f9c86beb2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-23 01:03:22.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/8"}}},"commit":{"parents":[{"commit":"81bfae6b9f045fe5433edaecbb556c4703c0b9bf","subject":"P9: Enable v2 API in devstack.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/81bfae6b9f045fe5433edaecbb556c4703c0b9bf"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-23 00:55:29.000000000","tz":540},"subject":"python3 patch (including zuul config)","message":"python3 patch (including zuul config)\n\n Fix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib.\n2nd: check \u0026 fix manually.\n    2-1: as_dict() function\n         fix as_dict() to work in py3 env.\n    2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n    2-3: fix all_pfs_have_vf() to be able to treat both list and iterator,\n         because filter() function returns list in py2, iterator in py3.\n    2-4: To treats map() in same way in py2/3 env , wrap map() with list()  in P6,P7\n    2-5: add python3.7 to setup.cfg\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\nStory: #2003219\nTask: #23418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0430c4bd7190dd161138e688bfd1293f9c86beb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0430c4bd7190dd161138e688bfd1293f9c86beb2"}]},"branch":"refs/heads/master"},"5f26f41c213bd2245dcec6808c7f1ea856409a19":{"kind":"REWORK","_number":9,"created":"2019-08-27 09:15:55.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/24/667524/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/9"}}},"commit":{"parents":[{"commit":"81bfae6b9f045fe5433edaecbb556c4703c0b9bf","subject":"P9: Enable v2 API in devstack.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/81bfae6b9f045fe5433edaecbb556c4703c0b9bf"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-27 09:13:33.000000000","tz":540},"subject":"python3 patch (including zuul config)","message":"python3 patch (including zuul config)\n\n Fix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib.\n2nd: check \u0026 fix manually.\n    2-1: as_dict() function\n         fix as_dict() to work in py3 env.\n    2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n    2-3: fix all_pfs_have_vf() to be able to treat both list and iterator,\n         because filter() function returns list in py2, iterator in py3.\n    2-4: To treats map() in same way in py2/3 env , wrap map() with list()  in P6,P7\n    2-5: add python3.7 to setup.cfg\n    2-6: add py3 tests, as voting\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\nStory: #2003219\nTask: #23418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5f26f41c213bd2245dcec6808c7f1ea856409a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5f26f41c213bd2245dcec6808c7f1ea856409a19"}]},"branch":"refs/heads/master"},"b3493526a4d594f934ece1c7a33715729543e89d":{"kind":"REWORK","_number":10,"created":"2019-09-10 00:43:43.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/24/667524/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/10"}}},"commit":{"parents":[{"commit":"dfb32d4866db697ed162b608865154c0e6682f3b","subject":"Notify Nova when binding finished","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dfb32d4866db697ed162b608865154c0e6682f3b"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-10 00:41:25.000000000","tz":-420},"subject":"python3 patch (including zuul config)","message":"python3 patch (including zuul config)\n\n Fix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib.\n2nd: check \u0026 fix manually.\n    2-1: as_dict() function\n         fix as_dict() to work in py3 env.\n    2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n    2-3: fix all_pfs_have_vf() to be able to treat both list and iterator,\n         because filter() function returns list in py2, iterator in py3.\n    2-4: To treats map() in same way in py2/3 env , wrap map() with list()  in P6,P7\n    2-5: add python3.7 to setup.cfg\n    2-6: add py3 tests, as voting\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\nStory: #2003219\nTask: #23418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b3493526a4d594f934ece1c7a33715729543e89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b3493526a4d594f934ece1c7a33715729543e89d"}]},"branch":"refs/heads/master"},"38119f675ea9e99a0045c63490243f1561c4cb9c":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-13 14:30:51.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/24/667524/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/667524/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/667524/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/667524/11"}}},"commit":{"parents":[{"commit":"2c50b9fd507f53a9fd745140cf86ee1d65756b54","subject":"Merge \"Notify Nova when all ARQs are resolved for an instance.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c50b9fd507f53a9fd745140cf86ee1d65756b54"}]}],"author":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-22 08:00:11.000000000","tz":540},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-13 16:27:22.000000000","tz":-420},"subject":"python3 patch (including zuul config)","message":"python3 patch (including zuul config)\n\n Fix points are extracted in following methods.\n\n1st: check with sixer:\n    change unicode function to use sixer lib.\n2nd: check \u0026 fix manually.\n    2-1: as_dict() function\n         fix as_dict() to work in py3 env.\n    2-2: delete unnecessary as_dict() call in response to the change in as_dict()\n    2-3: fix all_pfs_have_vf() to be able to treat both list and iterator,\n         because filter() function returns list in py2, iterator in py3.\n    2-4: To treats map() in same way in py2/3 env , wrap map() with list()  in P6,P7\n    2-5: add python3.7 to setup.cfg\n    2-6: add py3 tests, as voting\n\nChange-Id: Ibfdec7258c731ee45ce1b9bdba4b2024cea78a12\nStory: #2003219\nTask: #23418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/38119f675ea9e99a0045c63490243f1561c4cb9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/38119f675ea9e99a0045c63490243f1561c4cb9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
