)]}'
{"id":"openstack%2Fcyborg~668868","triplet_id":"openstack%2Fcyborg~master~Id190d2e12d842ec48782f4acfaa2cbd22cd4547a","project":"openstack/cyborg","branch":"master","topic":"driver","hashtags":[],"change_id":"Id190d2e12d842ec48782f4acfaa2cbd22cd4547a","subject":"Change attach_info, cpid_info from string to a dict.","status":"MERGED","created":"2019-07-03 11:23:40.000000000","updated":"2019-07-10 04:05:36.000000000","submitted":"2019-07-10 04:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668868-1562731536182-9c139318","meta_rev_id":"fa173fb0298fb541f3a485664cf9c107d35f9446","_number":668868,"virtual_id_number":668868,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-07-10 04:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"recommended":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"all":[{"value":1,"date":"2019-07-09 06:55:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":2,"date":"2019-07-10 02:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":1,"date":"2019-07-10 03:54:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 06:55:24.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-07-10 02:28:13.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-07-10 03:54:42.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-07-10 04:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c254797d532de1bd8e10167c7d4b4bdfc4e1a0c7","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-03 11:23:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0adf804746ad713d167aec4dca96c910dc12167","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:02:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/668868/1/check/openstack-tox-pep8/6fa1c0f/ : FAILURE in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/68/668868/1/check/openstack-tox-py27/dfd2ba1/ : FAILURE in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/68/668868/1/check/openstack-tox-docs/d42e860/html/ : SUCCESS in 3m 26s\n- cyborg-devstack http://logs.openstack.org/68/668868/1/check/cyborg-devstack/04080e3/ : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a62e34df339ad37a4ea1df70626fa9ff51573871","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-04 02:50:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38ee99fc93e6d4c71613abc5d9eedaaa3f472d98","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-04 03:03:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nThanks for the quick update, Xinran.","accounts_in_message":[],"_revision_number":2},{"id":"06498958d11f67ed531da12274b3e9f226e5dc26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 04:01:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/668868/2/check/openstack-tox-pep8/0440a1d/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/68/668868/2/check/openstack-tox-py27/8b46088/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/68/668868/2/check/openstack-tox-docs/13efeac/html/ : SUCCESS in 3m 04s\n- cyborg-devstack http://logs.openstack.org/68/668868/2/check/cyborg-devstack/bc5a3df/ : SUCCESS in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8a5dcce51e07eba4a78294bcd50824e47a4a841","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-04 08:51:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92ab11db1cf73d1280e7e8359e17fd21ec795928","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-04 08:58:40.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2383b8adc4274a636376772beb99313d807f03c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:30:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/668868/3/check/openstack-tox-pep8/48c14b3/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/68/668868/3/check/openstack-tox-py27/254e073/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/68/668868/3/check/openstack-tox-docs/8e157a1/html/ : SUCCESS in 3m 35s\n- cyborg-devstack http://logs.openstack.org/68/668868/3/check/cyborg-devstack/132e583/ : SUCCESS in 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"861cb85ed0956419d71556c0471d4af60d2dcefb","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-04 13:19:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nI still need to review the pf/vf usage.","accounts_in_message":[],"_revision_number":3},{"id":"47c0597417e993ce93df799fde826b938e156fa4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-04 13:25:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6fc8921eb0032d7a0fa39b317a480461e8a40fd4","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-05 02:44:55.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\nThanks Sundar, it is not exactly the same, especially for FPGA driver, please notice there is an \"if condition\" about pf_has_vf in FPGA driver but not others. If you want move it to generic driver, we can set this param to False by default if there is no objection from other driver developers. IMHO, I think we can merge this one first and the refactoring in another patch will be better.","accounts_in_message":[],"_revision_number":3},{"id":"bcc6e04dca7788e6dc4765bb6aef1dfac0252383","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-08 01:59:27.000000000","message":"Patch Set 3:\n\n\u003e (3 comments)\n \u003e \n \u003e I still need to review the pf/vf usage.\n\nHi Sundar, I have aligned with Yikun, he will rebase his patch after this patch merged.","accounts_in_message":[],"_revision_number":3},{"id":"4437a972ca10d4f1fb8f5e6b771f035f3fa0e3bb","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-08 02:04:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f2380d8d73e1540d6cc1d4edeb5b5b4608a07372","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-08 02:05:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"25e740a0b93264fda907cc6699db27381ec47364","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 02:42:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/668868/4/check/openstack-tox-pep8/8aa88cf/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/68/668868/4/check/openstack-tox-py27/b8f33e6/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/68/668868/4/check/openstack-tox-docs/e835b0d/html/ : SUCCESS in 3m 05s\n- cyborg-devstack http://logs.openstack.org/68/668868/4/check/cyborg-devstack/33661d1/ : SUCCESS in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6dddde887bafa1bbcb9ed48cb73d7272bdf95e70","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-08 06:20:27.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c657771d501ccc1d40713e919b58a77561ab559f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 06:58:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/668868/5/check/openstack-tox-pep8/1b29709/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/68/668868/5/check/openstack-tox-py27/07eabeb/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/68/668868/5/check/openstack-tox-docs/e41ac8c/html/ : SUCCESS in 4m 46s\n- cyborg-devstack http://logs.openstack.org/68/668868/5/check/cyborg-devstack/228488f/ : SUCCESS in 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7ef2e08ff28b40589eaf38c64e007405b5fc72a1","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-09 06:55:24.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"28152b42ca6fc70481f2f64493adac0d1356fb27","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-07-10 02:28:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dcae9e1105f6b93c3959d36e57bc8c969a3612b3","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-10 03:54:42.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4bdc9e8b2acb98cd208388f7b4c3056ea5142095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:54:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"20ec0e68905ea6e17a63bf1b34e0f2b743cdac39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 04:05:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/668868/5/gate/openstack-tox-pep8/698e87b/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/68/668868/5/gate/openstack-tox-py27/a833734/ : SUCCESS in 3m 07s\n- openstack-tox-docs http://logs.openstack.org/68/668868/5/gate/openstack-tox-docs/de7d31d/html/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":5},{"id":"4a81eaa8c2516588d8a31d0a25b55c7c59d71e4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 04:05:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9ca61345d8767be06c453850bf26870090bc6153","revisions":{"ae98546642a5ebd264a9bac16f4f43c07fa8e230":{"kind":"REWORK","_number":1,"created":"2019-07-03 11:23:40.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/68/668868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/68/668868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/68/668868/1"}}},"commit":{"parents":[{"commit":"553ca3bb0352de1cd6d7618277ad1f316504fdf7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/553ca3bb0352de1cd6d7618277ad1f316504fdf7"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-03 10:54:29.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-03 11:21:49.000000000","tz":480},"subject":"Change attach_info, cpid_info from sting to a dict.","message":"Change attach_info, cpid_info from sting to a dict.\n\nConsidering the field type has changed to DictofString, this patch will not\npass the gate test until all drivers are updated.\n\nPlease feel free to co-author.\n\nChange-Id: Id190d2e12d842ec48782f4acfaa2cbd22cd4547a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ae98546642a5ebd264a9bac16f4f43c07fa8e230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ae98546642a5ebd264a9bac16f4f43c07fa8e230"}]},"branch":"refs/heads/master"},"5ee2a76bcb6e1f40d149e8ec2ea7c057c9aff799":{"kind":"REWORK","_number":2,"created":"2019-07-04 02:50:24.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/68/668868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/68/668868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/68/668868/2"}}},"commit":{"parents":[{"commit":"553ca3bb0352de1cd6d7618277ad1f316504fdf7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/553ca3bb0352de1cd6d7618277ad1f316504fdf7"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-03 10:54:29.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-04 02:41:58.000000000","tz":480},"subject":"Change attach_info, cpid_info from sting to a dict.","message":"Change attach_info, cpid_info from sting to a dict.\n\nUpdate GPU and FPGA driver to match the current design:\ncpid_info and attach_handle info should be in JSON format.\n\nChange-Id: Id190d2e12d842ec48782f4acfaa2cbd22cd4547a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5ee2a76bcb6e1f40d149e8ec2ea7c057c9aff799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5ee2a76bcb6e1f40d149e8ec2ea7c057c9aff799"}]},"branch":"refs/heads/master"},"ba442cf7fdb6f724d65983f1f353dea41ad4c2af":{"kind":"REWORK","_number":3,"created":"2019-07-04 08:51:41.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/68/668868/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/68/668868/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/68/668868/3"}}},"commit":{"parents":[{"commit":"553ca3bb0352de1cd6d7618277ad1f316504fdf7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/553ca3bb0352de1cd6d7618277ad1f316504fdf7"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-03 10:54:29.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-04 08:51:15.000000000","tz":480},"subject":"Change attach_info, cpid_info from sting to a dict.","message":"Change attach_info, cpid_info from sting to a dict.\n\nUpdate GPU and FPGA driver to match the current design:\ncpid_info and attach_handle info should be in JSON format.\n\nChange-Id: Id190d2e12d842ec48782f4acfaa2cbd22cd4547a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ba442cf7fdb6f724d65983f1f353dea41ad4c2af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ba442cf7fdb6f724d65983f1f353dea41ad4c2af"}]},"branch":"refs/heads/master"},"8935ef42d5f6707aaeee98a8c03ba8de33703690":{"kind":"REWORK","_number":4,"created":"2019-07-08 02:04:49.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/68/668868/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/68/668868/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/68/668868/4"}}},"commit":{"parents":[{"commit":"553ca3bb0352de1cd6d7618277ad1f316504fdf7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/553ca3bb0352de1cd6d7618277ad1f316504fdf7"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-03 10:54:29.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-08 02:02:43.000000000","tz":480},"subject":"Change attach_info, cpid_info from sting to a dict.","message":"Change attach_info, cpid_info from sting to a dict.\n\nUpdate GPU and FPGA driver to match the current design:\ncpid_info and attach_handle info should be in JSON format.\n\nChange-Id: Id190d2e12d842ec48782f4acfaa2cbd22cd4547a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8935ef42d5f6707aaeee98a8c03ba8de33703690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8935ef42d5f6707aaeee98a8c03ba8de33703690"}]},"branch":"refs/heads/master"},"9ca61345d8767be06c453850bf26870090bc6153":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-08 06:20:27.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/68/668868/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/68/668868/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/68/668868/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/68/668868/5"}}},"commit":{"parents":[{"commit":"553ca3bb0352de1cd6d7618277ad1f316504fdf7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/553ca3bb0352de1cd6d7618277ad1f316504fdf7"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-07-03 10:54:29.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2019-07-08 06:20:22.000000000","tz":0},"subject":"Change attach_info, cpid_info from string to a dict.","message":"Change attach_info, cpid_info from string to a dict.\n\nUpdate GPU and FPGA driver to match the current design:\ncpid_info and attach_handle info should be in JSON format.\n\nChange-Id: Id190d2e12d842ec48782f4acfaa2cbd22cd4547a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9ca61345d8767be06c453850bf26870090bc6153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9ca61345d8767be06c453850bf26870090bc6153"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
