)]}'
{"id":"openstack%2Fcyborg~670466","triplet_id":"openstack%2Fcyborg~master~Ic73cd84cc1199ec150795a1a389eae73eda3555e","project":"openstack/cyborg","branch":"master","topic":"nova-integ","hashtags":[],"change_id":"Ic73cd84cc1199ec150795a1a389eae73eda3555e","subject":"P5: Basic changes for API layer.","status":"MERGED","created":"2019-07-12 08:15:00.000000000","updated":"2019-09-10 13:10:42.000000000","submitted":"2019-09-10 13:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670466-1568120940153-f0624132","meta_rev_id":"81ebcd1adeb742086cd1719514de449f4e5aa07f","_number":670466,"virtual_id_number":670466,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-09-10 13:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":-2,"date":"2019-09-10 13:10:42.000000000","post_submit":true,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-09-10 08:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2019-09-10 06:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-10 07:25:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":2,"date":"2019-09-10 07:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":1,"date":"2019-09-10 12:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 01:06:07.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2019-09-03 16:23:29.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-09-04 05:49:17.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-09-04 05:49:30.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2019-09-05 11:33:17.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-09-06 00:36:30.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-09-06 14:10:49.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-09-10 06:28:13.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-09-10 07:25:58.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2019-09-10 07:55:25.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-09-10 08:25:49.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-09-10 13:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3ab0bc139ef90c33abbb88dbf0f9250e9f02650","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 08:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"032551ffba05f3b722c2ebeaa60f8dbc244069c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 08:53:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/66/670466/1/check/openstack-tox-pep8/9a57f21/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/66/670466/1/check/openstack-tox-py27/da3654a/ : FAILURE in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/66/670466/1/check/openstack-tox-docs/8bad557/html/ : SUCCESS in 2m 45s\n- cyborg-devstack http://logs.openstack.org/66/670466/1/check/cyborg-devstack/122be19/ : SUCCESS in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06ea2bb9c106b03b31bdd03141b1e74f3ae8f7ec","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 09:11:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c06671ac05ea81c269bc757c75191a4f43e99e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 09:55:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/670466/2/check/openstack-tox-pep8/1486534/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/66/670466/2/check/openstack-tox-py27/ac1a076/ : SUCCESS in 2m 51s\n- openstack-tox-docs http://logs.openstack.org/66/670466/2/check/openstack-tox-docs/b231f90/html/ : SUCCESS in 2m 33s\n- cyborg-devstack http://logs.openstack.org/66/670466/2/check/cyborg-devstack/25abdc9/ : SUCCESS in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"388900f75f0b6cbd2c5f154f8446441957a8fc31","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 15:07:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2966da01cf0194be921ec3c86290bbd1d220790d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 15:48:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/670466/3/check/openstack-tox-pep8/c591715/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/66/670466/3/check/openstack-tox-py27/e28f07c/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/66/670466/3/check/openstack-tox-docs/2128b02/html/ : SUCCESS in 3m 25s\n- cyborg-devstack http://logs.openstack.org/66/670466/3/check/cyborg-devstack/032c23c/ : SUCCESS in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"761947f470245bd157169a14b6c07cfc10421f6a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-29 06:11:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"471d8d64e6d0ceea087620c7341cfa10a31cf479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 07:33:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/670466/4/check/openstack-tox-pep8/a4972f7/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/66/670466/4/check/openstack-tox-py27/1e5edf0/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/66/670466/4/check/openstack-tox-docs/e29a8c2/html/ : SUCCESS in 4m 57s\n- cyborg-devstack http://logs.openstack.org/66/670466/4/check/cyborg-devstack/1695bf5/ : FAILURE in 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2acc47dde6381370e947eae0109bbb8635eebc8","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-30 04:41:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"2baa8764d10be792bccc1d5953027d50078bcd83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 05:18:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/670466/5/check/openstack-tox-pep8/04a0824/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/66/670466/5/check/openstack-tox-py27/a8edbc1/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/66/670466/5/check/openstack-tox-docs/6706a03/html/ : SUCCESS in 3m 32s\n- cyborg-devstack http://logs.openstack.org/66/670466/5/check/cyborg-devstack/cb4c232/ : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6d9fa1ccdef6858c8c563778d4b1ebaba4651b02","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-31 07:26:55.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"015a74b9fd1803ea989c02c12667dae4a367672e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 08:05:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/670466/6/check/openstack-tox-pep8/5edcb49/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/66/670466/6/check/openstack-tox-py27/c625dee/ : SUCCESS in 3m 20s\n- openstack-tox-docs http://logs.openstack.org/66/670466/6/check/openstack-tox-docs/0a50eef/html/ : SUCCESS in 3m 03s\n- cyborg-devstack http://logs.openstack.org/66/670466/6/check/cyborg-devstack/8037152/ : SUCCESS in 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"589ac6a443a30041ae11e7143ebc9a8ffef93925","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-31 13:46:12.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nThanks Sundar, please see my comments inline.","accounts_in_message":[],"_revision_number":6},{"id":"b97ce3e91f27132e3e3c6a1e6015c0a2951a3108","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-08-01 16:22:34.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"60864ba0aed00b60d40579275c029be0c0da498c","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-08-05 03:21:09.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c313827893b80eaeb9c92d5939dd37bfbdc4d16e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 03:50:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/66/670466/7/check/openstack-tox-pep8/b87f9c3/ : SUCCESS in 3m 34s\n- openstack-tox-py27 https://logs.opendev.org/66/670466/7/check/openstack-tox-py27/6ef7f73/ : SUCCESS in 3m 21s\n- openstack-tox-docs https://logs.opendev.org/66/670466/7/check/openstack-tox-docs/2e2ea6c/html/ : SUCCESS in 3m 10s\n- cyborg-devstack https://logs.opendev.org/66/670466/7/check/cyborg-devstack/a11509a/ : SUCCESS in 27m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"24ad8dcb9fce5c503d60a31311f9f99d775efbc1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-05 06:11:37.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nThanks, Xinran and Shaohe.","accounts_in_message":[],"_revision_number":6},{"id":"4ff70ffca0a20e9232d244f4e2358c5e26f44adc","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-08-05 08:59:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bd71ee44be9911251ea41e9f0fca6dee85aa8ae3","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-05 20:46:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c6a542eb8c911060cc5acc2257b172ab929c724a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-05 20:48:47.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThanks, Shaohe.","accounts_in_message":[],"_revision_number":6},{"id":"07c4f958cd27b4a2fad18ffa7f5cc3f1b79cfbd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 21:17:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/66/670466/8/check/openstack-tox-pep8/46cac72/ : SUCCESS in 4m 23s\n- openstack-tox-py27 https://logs.opendev.org/66/670466/8/check/openstack-tox-py27/1979b7c/ : SUCCESS in 3m 33s\n- openstack-tox-docs https://logs.opendev.org/66/670466/8/check/openstack-tox-docs/33c4cc0/html/ : SUCCESS in 4m 49s\n- cyborg-devstack https://logs.opendev.org/66/670466/8/check/cyborg-devstack/afc3669/ : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7921810376d44707c882d14c2597d5851bf5434a","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-08-08 09:47:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"63d1c5dd97762c2e71a0446ed95c15ce03df226d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:28:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/66/670466/9/check/openstack-tox-pep8/011275b/ : SUCCESS in 4m 27s\n- openstack-tox-py27 https://logs.opendev.org/66/670466/9/check/openstack-tox-py27/a34b350/ : SUCCESS in 5m 11s\n- openstack-tox-docs https://logs.opendev.org/66/670466/9/check/openstack-tox-docs/b645d4e/html/ : SUCCESS in 4m 28s\n- cyborg-devstack https://logs.opendev.org/66/670466/9/check/cyborg-devstack/bbdf9f1/ : SUCCESS in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6b5583eda74c3ecdd08938b9a036d9c7b30594fb","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-08-08 10:42:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4b12a4bba042c08e1be05e2ae58f7d10f64ff81e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 11:17:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/66/670466/10/check/openstack-tox-pep8/9b6044b/ : SUCCESS in 3m 26s\n- openstack-tox-py27 https://logs.opendev.org/66/670466/10/check/openstack-tox-py27/b5271c2/ : SUCCESS in 4m 39s\n- openstack-tox-docs https://logs.opendev.org/66/670466/10/check/openstack-tox-docs/a30dc48/html/ : SUCCESS in 3m 19s\n- cyborg-devstack https://logs.opendev.org/66/670466/10/check/cyborg-devstack/414545c/ : SUCCESS in 30m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"737adfd458115966527101e89271ae36c34e05e4","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-08-08 11:28:46.000000000","message":"Topic changed from test to nova-integ","accounts_in_message":[],"_revision_number":10},{"id":"f2079975b25cd9b036ca873ce0e51403f32016b8","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 06:42:19.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"b9f3d816e8e1efdd5498f38fb9bc3d3e7673e832","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 06:47:02.000000000","message":"Patch Set 11: Code-Review-2\n\nProcedural -2 to prevent this from merging till later patches in this series have got +1/+2. This enables the whole series to merge together and to ensure that, if discussions of later patches prompt changes to previous patches, there is no problem.\n\nPlease review this patch and later ones as usual.","accounts_in_message":[],"_revision_number":11},{"id":"b028d6d4e93d12cd8e669db70b4b6cbe1a58d8f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 07:41:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/66/670466/11/check/openstack-tox-pep8/a3bc24a/ : SUCCESS in 4m 40s\n- openstack-tox-py27 https://logs.opendev.org/66/670466/11/check/openstack-tox-py27/3ea3a63/ : SUCCESS in 3m 20s\n- openstack-tox-docs https://logs.opendev.org/66/670466/11/check/openstack-tox-docs/1836b1b/html/ : SUCCESS in 4m 13s\n- cyborg-devstack https://logs.opendev.org/66/670466/11/check/cyborg-devstack/67c7b7a/ : SUCCESS in 46m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c63bc0df721f8d02b60d952c80ccb668fa83fbc0","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-12 05:07:37.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"971fec9e1a2bc07a14662f8179ad861d041ef738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 05:36:42.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/66/670466/12/check/openstack-tox-pep8/7ff20d6/ : SUCCESS in 3m 38s\n- openstack-tox-py27 https://logs.opendev.org/66/670466/12/check/openstack-tox-py27/8db4ed0/ : SUCCESS in 3m 47s\n- openstack-tox-docs https://logs.opendev.org/66/670466/12/check/openstack-tox-docs/672cd3f/html/ : SUCCESS in 3m 18s\n- cyborg-devstack https://logs.opendev.org/66/670466/12/check/cyborg-devstack/186507f/ : SUCCESS in 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"25a72a03acc8847d1bf6b8ba93acfea94781cf22","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-23 01:03:22.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"bd4d2a0f2fafa89fb355160044e6a29fca1e8481","author":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"date":"2019-08-23 01:06:07.000000000","message":"Patch Set 13:\n\nJust rebased on the gate issue patch[1].\n\n[1]https://review.opendev.org/#/c/677956/","accounts_in_message":[],"_revision_number":13},{"id":"6d8bff4141c815ebf10d3650ba9bc10d13bb011b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:41:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97a4c0144c541d28b3c5b46a04435b4 : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c802c3ba63f4787850f8c11f216529b : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc5a74c026d44188f46c83b36968dac : SUCCESS in 7m 01s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/d1fe5e3fc0d04036932e30bbd24f7f05 : SUCCESS in 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7abcbcdaee3c06db91b4504c882523024323d75a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-28 06:24:57.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"2274207df21e4273c75cab7c530c9712b0ef3983","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:40:31.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251a21a8e771405fb98f10879717d14c : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14abb6e0dc4b445ab84f950c6903fd00 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1444582c4049ddbfce01b7d87bea99 : SUCCESS in 5m 32s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/80cfbf90bcd64e5e9fb5eee7e145f954 : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b3bf90f57431b14ae562cc46fe64694252f6b34d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-29 03:31:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0f6963afa6bc197b316a1e9c7fe2f5759a10eeed","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-03 04:38:28.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"0cb374cf48de44511038eec8ab9a6b76d9623633","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 05:08:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d961fecc4f4d8bb9304b7a89443499 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad28e7cfbbc242ab9ee1c385f9e1a57c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde1b7063dc54187b129d55b61fb8b13 : SUCCESS in 4m 51s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/a4951568068f467facfef7d5568853bb : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"fc3f863f583d35dc9d62dfbeca9449ca9a7246ee","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-03 05:39:25.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"83b48bc1fbc1f4a759999f5ece1442433445920e","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-09-03 16:23:29.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"97d1301ea2b822b39514f99f97efbfcdd5f06fbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:08:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8dcb31038e476c878a00815b391684 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14be54f38b8e482593c5507e173afd4d : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3358d32a12864e23a148a354bbd0b3fb : SUCCESS in 5m 09s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/77fd4fd96dbe471b943d4bbc8bced962 : SUCCESS in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"dd89cf1fc3c4d6da19d10045a60844eb555e134e","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-04 03:41:56.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"cd8f5db9d46e64ba7c68f394168f3f9199f8d0bf","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-04 03:44:54.000000000","message":"Topic changed from nova-integ-patch18 to nova-integ","accounts_in_message":[],"_revision_number":17},{"id":"ceb569dabbb7550ad338f9253bb63fd1f008477a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 04:17:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779c1fc35743435283bec317141303ef : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/328af0ae54114cb3a6f6ef67d6a89da9 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ba8a809288497c8f8c3f076dd68ac4 : SUCCESS in 3m 43s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/0fac1c693e7f4c6ebc08774f9e9f528a : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0935564a56896e80e5b09cb3cc054b14171ad357","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-04 07:07:30.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"26d2751df19e9e31757f1186a25bb804b0ba87cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 07:39:11.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d8f380d91847a6af7a375df53fc6a0 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d704caa90dc1428b9923cb37a131011a : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27dcf0d22e694e69905ded29c122bf69 : SUCCESS in 6m 49s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/c35a701813874a5bb09256227710ffe5 : SUCCESS in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"76191ee3a914e307782d5774e29ad824485f95d4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-04 08:13:38.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"eb9a5faa655c2460093525e02881985e9da5dbad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 08:55:44.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a96d6d0f0043c88f0cc2fa54e94c26 : FAILURE in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e414bc0d4b4a4983a696e07c1ab77ec7 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05ae8b749bf429980e75a7df1ddbffb : SUCCESS in 4m 27s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/cb0c77cd5da24dc281705b7df71cdbe6 : SUCCESS in 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"7996aaba7ef01e6b32daa46442f24791df890fc0","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-04 16:11:30.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"795ee0946ec37cd65c1fdc62a1ebfa06f237673c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:00:19.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c57b657026a42a19d0fd2fe6f44a534 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/856c2c4e2a3e499886de7909bdc38937 : POST_FAILURE in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7fa987611d44adf8fe1914d01023c5d : POST_FAILURE in 4m 14s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/2f25801374b54f7ab05fbb7cd53ea397 : SUCCESS in 41m 05s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"854364a8d4a13e7bc01de46cf27405a0f8316e1c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-04 20:34:18.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"048f997a1be8c417ca081d0b39956ebf909abee9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:22:59.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9020de4ff1e14b26933caacbea3cbc54 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17bb5716bb0f48c194577ad27cd2ab3d : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1bc20672dd45008bc56da9f86e83ad : SUCCESS in 4m 52s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/ff5c2b9e7df546dd9ccb95931b2c3865 : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"bc8ec708e5dd9a073ddd4b71fd9172ebe31f870b","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-09-05 11:33:17.000000000","message":"Patch Set 20: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":20},{"id":"9b8a57ff42fcde190155c6dc917a37dbaa0e0b27","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-09-05 11:54:21.000000000","message":"Patch Set 20: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"110fe5a67b9139dd4c47f99a204f53d2d19daaaa","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-05 14:08:55.000000000","message":"Patch Set 20:\n\n(2 comments)\n\nThanks, Zhurong and Yikun.","accounts_in_message":[],"_revision_number":20},{"id":"86ae9471f323d37ffa2c7d156eecdce87a6bdce5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-05 19:14:13.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"57e45d478445a3590999f508159bd4e6697c2f5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:50:24.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b79e68d9554442d89748c981d3fcf4c : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7235375cb784eaf84e377043cd9a6c5 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3657d067c45a4cb885d992d081e7ae92 : SUCCESS in 5m 19s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/f9f93442671b4a7fad85928b2691d815 : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"146a8d8f4bea7aa68f010613eca79a1ae745bbf1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-09-06 00:36:30.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"6e8c7deb11b30b86017c4d81bcc6e8dfea3256d6","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-09-06 09:23:14.000000000","message":"Patch Set 21: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":21},{"id":"99794028b56c391e7ce299a4bd98a553b98c919f","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-09-06 09:35:50.000000000","message":"Patch Set 21: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":21},{"id":"0ecd3aae86201afe92e4ab52bd18a5eb550197c5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-06 11:16:19.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"6a302a7f4e2193d9a6ad5ac7d72c059224c8d4e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 11:53:02.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a043339fea4f9d856b13b47cc863c6 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52a0867914bd4854a75da580429d9454 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca7d3850b074e3997e00e4cee60c2d2 : SUCCESS in 4m 35s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/b8cdcd98b8254ffa91a0fa7b5c184959 : SUCCESS in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5eeae51ceb8243c441597a66ee4c482cb87d7213","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-06 13:37:32.000000000","message":"Patch Set 22:\n\n(4 comments)\n\nsome questions and comments within","accounts_in_message":[],"_revision_number":22},{"id":"97f02cb31ddc4afbb877ab1f4525f26d7a63b63a","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-06 14:10:49.000000000","message":"Patch Set 22:\n\nAs a style thing it would be great if you included actual information in your commit messages.","accounts_in_message":[],"_revision_number":22},{"id":"cf8e035048680ab2e32218030f1486bfae21eb79","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-07 23:40:39.000000000","message":"Patch Set 22:\n\n(4 comments)\n\nThanks a lot, Chris. :)\n\n \u003e As a style thing it would be great if you included actual\n \u003e information in your commit messages.\n\nDone.","accounts_in_message":[],"_revision_number":22},{"id":"a9f74dda01d228557035cd867534ef80aea9983e","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-08 17:18:28.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"c854c21aedc6a6ad034afea94eee34b159d1339d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 17:48:46.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788959ddbc7845728e5e80bb6db1ecf7 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81b9d8fd7ab74ca48cc48949f01dcc91 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1533d2ec812a4ac196dc56a15df3780b : SUCCESS in 4m 25s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/cfa52a861d1b4b7c84fa25e492867b50 : SUCCESS in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"3861043fd4396bc8ca244f8f519ce2c2676d0711","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-09 23:57:49.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"a6fba45a66e760e798b9f6ef948b824d968ca4a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 00:33:02.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20593a2d8c84b74a41936c492738cb9 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f79b3162c6b4d79a4f9a03c70f1ebf0 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45638dab00cd421bbd27235b260d91e9 : SUCCESS in 4m 04s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/43fbeed797b147cfbe7b936a385ebb80 : SUCCESS in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c33451aca84bcb8d5d0a42b51c7e16ca50067044","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 01:20:27.000000000","message":"Patch Set 24: -Code-Review Workflow-1\n\nRemoving -2 as we are close to the finish, hopefully.","accounts_in_message":[],"_revision_number":24},{"id":"6e0f45abfb378daaff998201b7a489c33320dd1e","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-09-10 06:28:13.000000000","message":"Patch Set 24: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":24},{"id":"b9b63571b0377e4d04b3e838fafb3355b76b6f5c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-09-10 07:25:58.000000000","message":"Patch Set 24: Code-Review+1","accounts_in_message":[],"_revision_number":24},{"id":"386cfa3bd44b56b8c42f968e7132ee67ea0c478a","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-09-10 07:55:25.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"967d59580deaafcfa18d2af0138be63d4b1d4865","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-09-10 08:25:49.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"fc15b8c01dd4db9d55076000090b61862af07e9a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 12:58:20.000000000","message":"Patch Set 24: Workflow+1\n\nThank you all for the timely reviews!","accounts_in_message":[],"_revision_number":24},{"id":"e85ba84d8d2e7e513926a4f26364db8f39b0cac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:58:32.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"4d92c67638d718deccf4594f01380d251054ed2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:08:59.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f5b80bbd5a46ae80c81431d1c479c2 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/397c6ea80f494480b0f00e3b3bbecddd : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a442e33c507241d7b896e3a13e8dac54 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":24},{"id":"63f61285d6d6492b0121525c85eee744cb24d44a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:09:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":24},{"id":"b25ae17fe5971cd8a543b31733c96e43b32538c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:10:42.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f1281734fed405e9d443f3bf31161b0 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"2ec1a8715f23acbdb38b04e39eabc7a80e19fa2b","revisions":{"1f59eca1868522af79220c98360ff80aaae3ee9a":{"kind":"REWORK","_number":1,"created":"2019-07-12 08:15:00.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/1"}}},"commit":{"parents":[{"commit":"345a9ebb2dafe9e42b37fd8dbf12b4ab63a1c018","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/345a9ebb2dafe9e42b37fd8dbf12b4ab63a1c018"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 08:05:33.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1f59eca1868522af79220c98360ff80aaae3ee9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1f59eca1868522af79220c98360ff80aaae3ee9a"}]},"branch":"refs/heads/master"},"6e0ad7284fea299054ece45106cef83ccc7d9666":{"kind":"REWORK","_number":2,"created":"2019-07-12 09:11:49.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/2"}}},"commit":{"parents":[{"commit":"345a9ebb2dafe9e42b37fd8dbf12b4ab63a1c018","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/345a9ebb2dafe9e42b37fd8dbf12b4ab63a1c018"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 09:05:07.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6e0ad7284fea299054ece45106cef83ccc7d9666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6e0ad7284fea299054ece45106cef83ccc7d9666"}]},"branch":"refs/heads/master"},"7b43c4ba02a8201f7f72e7d36de0a0d63af5d031":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-12 15:07:04.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/3"}}},"commit":{"parents":[{"commit":"5b2118f1b28b25811a082a72bcdb5d96e09abedb","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b2118f1b28b25811a082a72bcdb5d96e09abedb"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 15:05:25.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7b43c4ba02a8201f7f72e7d36de0a0d63af5d031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7b43c4ba02a8201f7f72e7d36de0a0d63af5d031"}]},"branch":"refs/heads/master"},"63a2964e084bba110f13c230f8ef0d547e81c607":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-29 06:11:31.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/4"}}},"commit":{"parents":[{"commit":"35c27df9c5b9f834d10f8ab785c436359d4e29c6","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/35c27df9c5b9f834d10f8ab785c436359d4e29c6"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-29 06:09:36.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63a2964e084bba110f13c230f8ef0d547e81c607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63a2964e084bba110f13c230f8ef0d547e81c607"}]},"branch":"refs/heads/master"},"d763791b819c621efba2003c94ccfb9794d5dbbc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-30 04:41:04.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/5"}}},"commit":{"parents":[{"commit":"6a0c909576ffd76b3dd90ffcccf6d1ea504f6fa5","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6a0c909576ffd76b3dd90ffcccf6d1ea504f6fa5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-30 04:38:28.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d763791b819c621efba2003c94ccfb9794d5dbbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d763791b819c621efba2003c94ccfb9794d5dbbc"}]},"branch":"refs/heads/master"},"b9cfe45c6f6d1718a6411009e2b48bf7a0fd4b77":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-31 07:26:55.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/6"}}},"commit":{"parents":[{"commit":"be36ee9d2da5aa521626a210231777c5faf6291b","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/be36ee9d2da5aa521626a210231777c5faf6291b"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-31 07:21:12.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b9cfe45c6f6d1718a6411009e2b48bf7a0fd4b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b9cfe45c6f6d1718a6411009e2b48bf7a0fd4b77"}]},"branch":"refs/heads/master"},"0c746299852777272c58697213e147214966c793":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-05 03:21:09.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/66/670466/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/7"}}},"commit":{"parents":[{"commit":"0d99c026c475339127df70da38e3db418058a949","subject":"Merge \"Mailing lists change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0d99c026c475339127df70da38e3db418058a949"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2019-08-05 03:20:22.000000000","tz":0},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0c746299852777272c58697213e147214966c793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0c746299852777272c58697213e147214966c793"}]},"branch":"refs/heads/master"},"246a46595ec017276315f58ddf60e42fdb8f04e6":{"kind":"REWORK","_number":8,"created":"2019-08-05 20:46:26.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/8"}}},"commit":{"parents":[{"commit":"be36ee9d2da5aa521626a210231777c5faf6291b","subject":"P4: Basic changes for objects layer.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/be36ee9d2da5aa521626a210231777c5faf6291b"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-05 18:59:37.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/246a46595ec017276315f58ddf60e42fdb8f04e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/246a46595ec017276315f58ddf60e42fdb8f04e6"}]},"branch":"refs/heads/master"},"f85b83f6db39e24ffd0b0ac4e369328ee5ae582c":{"kind":"REWORK","_number":9,"created":"2019-08-08 09:47:58.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/66/670466/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/9"}}},"commit":{"parents":[{"commit":"6d79037f2131ba141f2533f37220bd99dd6c1746","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d79037f2131ba141f2533f37220bd99dd6c1746"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2019-08-08 09:45:05.000000000","tz":0},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f85b83f6db39e24ffd0b0ac4e369328ee5ae582c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f85b83f6db39e24ffd0b0ac4e369328ee5ae582c"}]},"branch":"refs/heads/master"},"d0e44730173bef0b5439d333d27b7a0146c1d24d":{"kind":"REWORK","_number":10,"created":"2019-08-08 10:42:26.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/66/670466/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/10"}}},"commit":{"parents":[{"commit":"6d79037f2131ba141f2533f37220bd99dd6c1746","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d79037f2131ba141f2533f37220bd99dd6c1746"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2019-08-08 10:30:29.000000000","tz":0},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d0e44730173bef0b5439d333d27b7a0146c1d24d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d0e44730173bef0b5439d333d27b7a0146c1d24d"}]},"branch":"refs/heads/master"},"4c7983f65d184e6fcc05b243e079cfd7dfc6ad81":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-08-09 06:42:19.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/11"}}},"commit":{"parents":[{"commit":"0d99c026c475339127df70da38e3db418058a949","subject":"Merge \"Mailing lists change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0d99c026c475339127df70da38e3db418058a949"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-09 05:49:17.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4c7983f65d184e6fcc05b243e079cfd7dfc6ad81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4c7983f65d184e6fcc05b243e079cfd7dfc6ad81"}]},"branch":"refs/heads/master"},"4e541c49f1e14209f1aa38cfb51d6f0329b2b6f4":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-08-12 05:07:37.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/12"}}},"commit":{"parents":[{"commit":"8c26ba469667c1c86d8bf0b31c55cf79f35151b1","subject":"Merge \"Add the fake driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8c26ba469667c1c86d8bf0b31c55cf79f35151b1"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-12 05:01:44.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4e541c49f1e14209f1aa38cfb51d6f0329b2b6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4e541c49f1e14209f1aa38cfb51d6f0329b2b6f4"}]},"branch":"refs/heads/master"},"d0975138944990c69d83bdec3638ad1089098fef":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-08-23 01:03:22.000000000","uploader":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"ref":"refs/changes/66/670466/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/13"}}},"commit":{"parents":[{"commit":"cbcb434ace2c54a2cbc174e2e01d9dfe4d115578","subject":"Fix docs gate issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cbcb434ace2c54a2cbc174e2e01d9dfe4d115578"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","date":"2019-08-23 00:55:28.000000000","tz":540},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d0975138944990c69d83bdec3638ad1089098fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d0975138944990c69d83bdec3638ad1089098fef"}]},"branch":"refs/heads/master"},"3f088c37222ceabb82636e06c37b7c8523777ace":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-08-28 06:24:57.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/14"}}},"commit":{"parents":[{"commit":"fcdb49561989bd7075ef0c9a6121678e833adfc5","subject":"Merge \"fake driver bug fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fcdb49561989bd7075ef0c9a6121678e833adfc5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-28 02:29:18.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3f088c37222ceabb82636e06c37b7c8523777ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3f088c37222ceabb82636e06c37b7c8523777ace"}]},"branch":"refs/heads/master"},"fbfc9b226e12c7372717e78848933e8698e2c1e3":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-09-03 04:38:28.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/15"}}},"commit":{"parents":[{"commit":"af7e412a81d5722d51d02c450174a9480cafbce2","subject":"Merge \"Move to releases.openstack.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/af7e412a81d5722d51d02c450174a9480cafbce2"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-02 05:51:44.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fbfc9b226e12c7372717e78848933e8698e2c1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fbfc9b226e12c7372717e78848933e8698e2c1e3"}]},"branch":"refs/heads/master"},"2c876518b8dc00518fe052a0cba93ac2a4c4d878":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-09-03 16:23:29.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/66/670466/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/16"}}},"commit":{"parents":[{"commit":"cda15a3f2e46a08abd2adc4031f41141e023e57d","subject":"Merge \"Fix pep8 error in cyborg/*.py and add Forbidden class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cda15a3f2e46a08abd2adc4031f41141e023e57d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"wangzh21","email":"wangzh21@lenovo.com","date":"2019-09-03 16:19:25.000000000","tz":480},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c876518b8dc00518fe052a0cba93ac2a4c4d878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c876518b8dc00518fe052a0cba93ac2a4c4d878"}]},"branch":"refs/heads/master"},"db6df1b1dff0392db8c7bc54eeededaa59c574cb":{"kind":"REWORK","_number":17,"created":"2019-09-04 03:41:56.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/17"}}},"commit":{"parents":[{"commit":"cda15a3f2e46a08abd2adc4031f41141e023e57d","subject":"Merge \"Fix pep8 error in cyborg/*.py and add Forbidden class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cda15a3f2e46a08abd2adc4031f41141e023e57d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-04 03:29:18.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/db6df1b1dff0392db8c7bc54eeededaa59c574cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/db6df1b1dff0392db8c7bc54eeededaa59c574cb"}]},"branch":"refs/heads/master"},"c9e1f3c7cd1efc542bcfd70adbfb45191f5e39e5":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-09-04 07:07:30.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/18"}}},"commit":{"parents":[{"commit":"2503fbb0fd1e4d0ed51c9729ce34d55519abc021","subject":"Merge \"bug fix: generate attach_handle query in DB session\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2503fbb0fd1e4d0ed51c9729ce34d55519abc021"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-04 06:49:43.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9e1f3c7cd1efc542bcfd70adbfb45191f5e39e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9e1f3c7cd1efc542bcfd70adbfb45191f5e39e5"}]},"branch":"refs/heads/master"},"03d6b17f1a865b662ab0dd6826c84cd9839bf8b1":{"kind":"REWORK","_number":19,"created":"2019-09-04 08:13:38.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/19"}}},"commit":{"parents":[{"commit":"0f87dca1242b251394e487f06cf81a418fef8530","subject":"Merge \"P4: Fix pep8 error in cyborg/api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0f87dca1242b251394e487f06cf81a418fef8530"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-04 08:06:40.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/03d6b17f1a865b662ab0dd6826c84cd9839bf8b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/03d6b17f1a865b662ab0dd6826c84cd9839bf8b1"}]},"branch":"refs/heads/master"},"130c53502f262a81073fab8a77cd0109da4f9f49":{"kind":"REWORK","_number":20,"created":"2019-09-04 16:11:30.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/20"}}},"commit":{"parents":[{"commit":"07efd624b8f957a9288f5d833550a1a93b804b5d","subject":"Placement report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07efd624b8f957a9288f5d833550a1a93b804b5d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-04 15:40:40.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/130c53502f262a81073fab8a77cd0109da4f9f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/130c53502f262a81073fab8a77cd0109da4f9f49"}]},"branch":"refs/heads/master"},"fe1ec8b92bd457b15a198a67736cfadc07f44b4c":{"kind":"REWORK","_number":21,"created":"2019-09-05 19:14:13.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/21"}}},"commit":{"parents":[{"commit":"07efd624b8f957a9288f5d833550a1a93b804b5d","subject":"Placement report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07efd624b8f957a9288f5d833550a1a93b804b5d"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-05 18:59:50.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fe1ec8b92bd457b15a198a67736cfadc07f44b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fe1ec8b92bd457b15a198a67736cfadc07f44b4c"}]},"branch":"refs/heads/master"},"df2201e76e892c035cfa032275e5a634a43a0c99":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-09-06 11:16:19.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/22"}}},"commit":{"parents":[{"commit":"f8c96e156fce563104d26d93470187e4e0e91ed6","subject":"Remove the delete rc_fields in fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f8c96e156fce563104d26d93470187e4e0e91ed6"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-06 11:09:25.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/df2201e76e892c035cfa032275e5a634a43a0c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/df2201e76e892c035cfa032275e5a634a43a0c99"}]},"branch":"refs/heads/master"},"133b08f6882380bdf9cf91d4bc7cca783937d825":{"kind":"REWORK","_number":23,"created":"2019-09-08 17:18:28.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/23"}}},"commit":{"parents":[{"commit":"4a28a7a96e1244a52282323ebdd9c4347a1e4d64","subject":"Merge \"Fix placement report error in python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4a28a7a96e1244a52282323ebdd9c4347a1e4d64"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-07 21:29:05.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nIntroduce changes to enable v2 APIs, needed for Nova integ.\n\n. Create v2/ dir under cyborg/api/controllers/.\n. Refactor common files and code in cyborg/api/controllers\n  to share between v1 and v2.\n. Enable service version discovery and microversions\n  by following [1].\n\n[1] https://docs.openstack.org/api-guide/compute/microversions.html\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/133b08f6882380bdf9cf91d4bc7cca783937d825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/133b08f6882380bdf9cf91d4bc7cca783937d825"}]},"branch":"refs/heads/master"},"2ec1a8715f23acbdb38b04e39eabc7a80e19fa2b":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-09-09 23:57:49.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/66/670466/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/66/670466/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/66/670466/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/66/670466/24"}}},"commit":{"parents":[{"commit":"ac39c5f228e1a5229003a373f74739ce728043fe","subject":"Merge \"Using str replace encode or decode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ac39c5f228e1a5229003a373f74739ce728043fe"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 07:29:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-09 23:51:11.000000000","tz":-420},"subject":"P5: Basic changes for API layer.","message":"P5: Basic changes for API layer.\n\nIntroduce changes to enable v2 APIs, needed for Nova integ.\n\n. Create v2/ dir under cyborg/api/controllers/.\n. Refactor common files and code in cyborg/api/controllers\n  to share between v1 and v2.\n. Enable service version discovery and microversions\n  by following [1].\n\n[1] https://docs.openstack.org/api-guide/compute/microversions.html\n\nChange-Id: Ic73cd84cc1199ec150795a1a389eae73eda3555e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2ec1a8715f23acbdb38b04e39eabc7a80e19fa2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2ec1a8715f23acbdb38b04e39eabc7a80e19fa2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
