)]}'
{"id":"openstack%2Fcyborg~674212","triplet_id":"openstack%2Fcyborg~master~I5ae38d24a4746006a24f93222ecccc58896bedce","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"I5ae38d24a4746006a24f93222ecccc58896bedce","subject":"TEST","status":"ABANDONED","created":"2019-08-02 07:12:30.000000000","updated":"2019-08-06 06:28:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dc11abe618619c56a31e208833155582e31f89c","_number":674212,"virtual_id_number":674212,"owner":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":-1,"date":"2019-08-02 09:31:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":-1,"date":"2019-08-02 08:56:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 09:31:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c0b695283521af6b101c2010b709b62a2066e16","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-08-02 07:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1224eb95a89057e17c4cf363eb0bf96093cf543d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-02 07:14:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"63d7dd81e7a762c1f6aef565880d69c05f3e9ff4","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-02 07:15:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2a5c02fdb576fd77798a6b5c78ee40f2d698f70b","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-08-02 07:24:42.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"f1f196868088c1f331668d727b1275fd6795fc66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 07:53:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/12/674212/3/check/openstack-tox-pep8/a9d6e60/ : FAILURE in 3m 37s\n- openstack-tox-py27 https://logs.opendev.org/12/674212/3/check/openstack-tox-py27/124777c/ : FAILURE in 4m 05s\n- requirements-check https://logs.opendev.org/12/674212/3/check/requirements-check/30d679e/ : FAILURE in 3m 35s\n- build-openstack-releasenotes https://logs.opendev.org/12/674212/3/check/build-openstack-releasenotes/5632914/html/ : SUCCESS in 3m 15s\n- openstack-tox-docs https://logs.opendev.org/12/674212/3/check/openstack-tox-docs/beecfa6/html/ : SUCCESS in 3m 20s\n- cyborg-devstack https://logs.opendev.org/12/674212/3/check/cyborg-devstack/0c0872a/ : SUCCESS in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2e45c3177e071bcc1415c52bf0d30add60f71c8e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-02 08:10:14.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"dc99ef6bc2bcf536acff537e59d96a4e1b7f0d7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 08:49:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/12/674212/4/check/test-release-openstack/408f83d/ : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://logs.opendev.org/12/674212/4/check/openstack-tox-pep8/a138221/ : FAILURE in 4m 25s\n- openstack-tox-py27 https://logs.opendev.org/12/674212/4/check/openstack-tox-py27/944911c/ : FAILURE in 3m 45s\n- requirements-check https://logs.opendev.org/12/674212/4/check/requirements-check/0e5ab56/ : FAILURE in 2m 30s\n- build-openstack-releasenotes https://logs.opendev.org/12/674212/4/check/build-openstack-releasenotes/4f94c37/html/ : SUCCESS in 3m 27s\n- openstack-tox-docs https://logs.opendev.org/12/674212/4/check/openstack-tox-docs/91a1515/html/ : SUCCESS in 4m 54s\n- cyborg-devstack https://logs.opendev.org/12/674212/4/check/cyborg-devstack/6010b55/ : SUCCESS in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7277ce8af2159985d54418e33653bb3e9154638f","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-02 08:51:10.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7f4f314b930a2d3511cff1133d092d3701670e4a","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-08-02 08:56:31.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"3b685053f1c74b311b02e454dd920a7b958fe2a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 09:31:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/12/674212/5/check/test-release-openstack/55e2387/ : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://logs.opendev.org/12/674212/5/check/openstack-tox-pep8/8b787ab/ : FAILURE in 3m 56s\n- openstack-tox-py27 https://logs.opendev.org/12/674212/5/check/openstack-tox-py27/3bfa31d/ : FAILURE in 4m 12s\n- requirements-check https://logs.opendev.org/12/674212/5/check/requirements-check/d5a18da/ : FAILURE in 3m 13s\n- build-openstack-releasenotes https://logs.opendev.org/12/674212/5/check/build-openstack-releasenotes/223dd29/html/ : SUCCESS in 3m 25s\n- openstack-tox-docs https://logs.opendev.org/12/674212/5/check/openstack-tox-docs/2802963/html/ : SUCCESS in 3m 05s\n- cyborg-devstack https://logs.opendev.org/12/674212/5/check/cyborg-devstack/c43e657/ : SUCCESS in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3dc11abe618619c56a31e208833155582e31f89c","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-08-06 06:28:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c9742e46bbeb0c5e0fe182ed3bb44e185d443313","revisions":{"2374cc141acb8115270c5aa914ea591f00187f16":{"kind":"REWORK","_number":1,"created":"2019-08-02 07:12:30.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/12/674212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/12/674212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/12/674212/1"}}},"commit":{"parents":[{"commit":"fa5fe26038ed6ded7fabe2363753dcf953c3308e","subject":"Merge \"P4: Basic changes for objects layer.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fa5fe26038ed6ded7fabe2363753dcf953c3308e"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-08-02 07:11:47.000000000","tz":-420},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-08-02 07:12:20.000000000","tz":-420},"subject":"TEST","message":"TEST\n\nChange-Id: I5ae38d24a4746006a24f93222ecccc58896bedce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2374cc141acb8115270c5aa914ea591f00187f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2374cc141acb8115270c5aa914ea591f00187f16"}]},"branch":"refs/heads/master"},"de876322c2090b48f035894da76bf7759bc3b039":{"kind":"REWORK","_number":2,"created":"2019-08-02 07:14:12.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/12/674212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/12/674212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/12/674212/2"}}},"commit":{"parents":[{"commit":"fa5fe26038ed6ded7fabe2363753dcf953c3308e","subject":"Merge \"P4: Basic changes for objects layer.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fa5fe26038ed6ded7fabe2363753dcf953c3308e"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-08-02 07:11:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-02 07:14:08.000000000","tz":0},"subject":"TEST","message":"TEST\n\nChange-Id: I5ae38d24a4746006a24f93222ecccc58896bedce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/de876322c2090b48f035894da76bf7759bc3b039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/de876322c2090b48f035894da76bf7759bc3b039"}]},"branch":"refs/heads/master"},"604ac96e5f3694190bab08d0510814af27b2399a":{"kind":"REWORK","_number":3,"created":"2019-08-02 07:15:40.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/12/674212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/12/674212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/12/674212/3"}}},"commit":{"parents":[{"commit":"fa5fe26038ed6ded7fabe2363753dcf953c3308e","subject":"Merge \"P4: Basic changes for objects layer.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fa5fe26038ed6ded7fabe2363753dcf953c3308e"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-08-02 07:11:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-02 07:15:36.000000000","tz":0},"subject":"TEST","message":"TEST\n\nChange-Id: I5ae38d24a4746006a24f93222ecccc58896bedce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/604ac96e5f3694190bab08d0510814af27b2399a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/604ac96e5f3694190bab08d0510814af27b2399a"}]},"branch":"refs/heads/master"},"427005b795d57d4d614336450eb072d6b8825c08":{"kind":"REWORK","_number":4,"created":"2019-08-02 08:10:14.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/12/674212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/12/674212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/12/674212/4"}}},"commit":{"parents":[{"commit":"fa5fe26038ed6ded7fabe2363753dcf953c3308e","subject":"Merge \"P4: Basic changes for objects layer.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fa5fe26038ed6ded7fabe2363753dcf953c3308e"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-08-02 07:11:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-02 08:10:06.000000000","tz":0},"subject":"TEST","message":"TEST\n\nChange-Id: I5ae38d24a4746006a24f93222ecccc58896bedce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/427005b795d57d4d614336450eb072d6b8825c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/427005b795d57d4d614336450eb072d6b8825c08"}]},"branch":"refs/heads/master"},"c9742e46bbeb0c5e0fe182ed3bb44e185d443313":{"kind":"REWORK","_number":5,"created":"2019-08-02 08:51:10.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/12/674212/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/12/674212/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/674212/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/12/674212/5"}}},"commit":{"parents":[{"commit":"fa5fe26038ed6ded7fabe2363753dcf953c3308e","subject":"Merge \"P4: Basic changes for objects layer.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fa5fe26038ed6ded7fabe2363753dcf953c3308e"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-08-02 07:11:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-02 08:51:06.000000000","tz":0},"subject":"TEST","message":"TEST\n\nChange-Id: I5ae38d24a4746006a24f93222ecccc58896bedce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9742e46bbeb0c5e0fe182ed3bb44e185d443313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9742e46bbeb0c5e0fe182ed3bb44e185d443313"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
