)]}'
{"id":"openstack%2Fcyborg~675058","triplet_id":"openstack%2Fcyborg~master~I7f2a65052e21bbb635e7c7c77b4c17ee6a7a6026","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"I7f2a65052e21bbb635e7c7c77b4c17ee6a7a6026","subject":"Bug-fix in FPGA driver test case.","status":"MERGED","created":"2019-08-07 08:36:25.000000000","updated":"2019-08-09 07:20:23.000000000","submitted":"2019-08-09 07:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675058-1565335223055-eb0b3e6e","meta_rev_id":"b3f3b66ef9de2ac720e09f91dafc9cf3acda3610","_number":675058,"virtual_id_number":675058,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"date":"2019-08-09 06:27:16.000000000","_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-08-09 07:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":2,"date":"2019-08-09 07:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":2,"date":"2019-08-09 06:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"date":"2019-08-07 08:38:27.000000000","_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":1,"date":"2019-08-09 07:10:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 08:38:27.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-08-09 06:33:37.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-08-09 07:10:20.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-08-09 07:20:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1913a0c32daa6c486ae1c4092005e1c90463e6fd","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-08-07 08:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f42c2843bd75866fdbf66dc0e03a25f149d4368","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 09:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/58/675058/1/check/openstack-tox-pep8/61b6aec/ : SUCCESS in 3m 57s\n- openstack-tox-py27 https://logs.opendev.org/58/675058/1/check/openstack-tox-py27/1a83778/ : SUCCESS in 4m 27s\n- openstack-tox-docs https://logs.opendev.org/58/675058/1/check/openstack-tox-docs/43b0cef/html/ : SUCCESS in 3m 59s\n- cyborg-devstack https://logs.opendev.org/58/675058/1/check/cyborg-devstack/dced7f5/ : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9af7f99d861b02bf2baaf79440fb2aea91c59d87","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 05:10:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7d8ee62ac5575b8190804560025faacabfbad35","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-08-09 06:27:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5d742ea2a34d99113c71657624f4815678c8b10","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 06:33:37.000000000","message":"Patch Set 1: Code-Review+2\n\nOk.","accounts_in_message":[],"_revision_number":1},{"id":"9613eed4060aa8a693991dc79adc14c889411688","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-08-09 07:10:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood for FPGA driver","accounts_in_message":[],"_revision_number":1},{"id":"ec10fd69e697ed05d077ee86fdc4e1c37c4a7c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 07:10:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d5dd3f1a8eab8999adcee46c38fdea60e66811c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 07:20:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/58/675058/1/gate/openstack-tox-pep8/4cb3d4f/ : SUCCESS in 4m 08s\n- openstack-tox-py27 https://logs.opendev.org/58/675058/1/gate/openstack-tox-py27/6ff1363/ : SUCCESS in 3m 28s\n- openstack-tox-docs https://logs.opendev.org/58/675058/1/gate/openstack-tox-docs/cec3053/html/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b3f3b66ef9de2ac720e09f91dafc9cf3acda3610","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 07:20:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03fb1e5b7cdc2ec2987c2cf0f174a04a9b2ebc8d","revisions":{"03fb1e5b7cdc2ec2987c2cf0f174a04a9b2ebc8d":{"kind":"REWORK","_number":1,"created":"2019-08-07 08:36:25.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/58/675058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/58/675058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/58/675058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/58/675058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/58/675058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/58/675058/1"}}},"commit":{"parents":[{"commit":"6d79037f2131ba141f2533f37220bd99dd6c1746","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d79037f2131ba141f2533f37220bd99dd6c1746"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2019-08-07 08:31:20.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2019-08-07 08:31:20.000000000","tz":0},"subject":"Bug-fix in FPGA driver test case.","message":"Bug-fix in FPGA driver test case.\n\nCurrently when we run the FPGA test cases on a host where real FPGA\nlocated, the test will not passed because the discover() return real\nFPGA info and fake FPGA info. In that case, the assertion will fail.\n\nThis patch fixed this.\n\nChange-Id: I7f2a65052e21bbb635e7c7c77b4c17ee6a7a6026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/03fb1e5b7cdc2ec2987c2cf0f174a04a9b2ebc8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/03fb1e5b7cdc2ec2987c2cf0f174a04a9b2ebc8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
