)]}'
{"id":"openstack%2Fcyborg~675535","triplet_id":"openstack%2Fcyborg~master~I64ce824f27838650f350c6138ef497354f49941d","project":"openstack/cyborg","branch":"master","topic":"nova-integ","hashtags":[],"change_id":"I64ce824f27838650f350c6138ef497354f49941d","subject":"P8.5: Remove NotImplementedError for some dbapi methods.","status":"ABANDONED","created":"2019-08-09 06:42:19.000000000","updated":"2019-08-19 06:42:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dab612ffd584b6ae659cf875a3cb451faea55152","_number":675535,"virtual_id_number":675535,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2019-08-12 05:43:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":-1,"date":"2019-08-19 06:41:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 05:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"192794f0bbaceff1eced2c52ab16f2a722191c70","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-09 06:42:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"663e2de6eede2f631d5323d38dab2c43cd2c00bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 07:37:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/35/675535/1/check/openstack-tox-pep8/f5c5ffd/ : FAILURE in 3m 13s\n- openstack-tox-py27 https://logs.opendev.org/35/675535/1/check/openstack-tox-py27/f497d06/ : FAILURE in 4m 35s\n- openstack-tox-docs https://logs.opendev.org/35/675535/1/check/openstack-tox-docs/b59fb65/html/ : SUCCESS in 4m 06s\n- cyborg-devstack https://logs.opendev.org/35/675535/1/check/cyborg-devstack/422b9ea/ : SUCCESS in 27m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea6c1122813a91c5ab465d70f98e87a19aca2e6e","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-12 05:07:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4cf7c11cba3c58146c0c0cbde7761fddfca67fc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 05:43:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/35/675535/2/check/openstack-tox-pep8/75928af/ : SUCCESS in 4m 04s\n- openstack-tox-py27 https://logs.opendev.org/35/675535/2/check/openstack-tox-py27/65dbd37/ : SUCCESS in 4m 24s\n- openstack-tox-docs https://logs.opendev.org/35/675535/2/check/openstack-tox-docs/36de238/html/ : SUCCESS in 3m 42s\n- cyborg-devstack https://logs.opendev.org/35/675535/2/check/cyborg-devstack/8edcc96/ : SUCCESS in 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f98ab1b6ed51f7e9164c068be77be9459c5e4221","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-19 06:41:52.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b43f35d8d31e949e7f3572fb6a2d951b43caeb48","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-19 06:42:46.000000000","message":"Abandoned\n\nMerged into P5.5.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d8754a9fc73ef53d1b43fc9c01f85cf009264990","revisions":{"f1a361bd83473a8bea6240bc8257eb9a24f7870b":{"kind":"REWORK","_number":1,"created":"2019-08-09 06:42:19.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/35/675535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/35/675535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/35/675535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/35/675535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/35/675535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/35/675535/1"}}},"commit":{"parents":[{"commit":"7ba226334775b097dbd531492997cd2d30cbe99f","subject":"P8: Changes for programming.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7ba226334775b097dbd531492997cd2d30cbe99f"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-09 05:56:36.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-09 05:56:36.000000000","tz":-420},"subject":"P8.5: Remove NotImplementedError for some dbapi methods.","message":"P8.5: Remove NotImplementedError for some dbapi methods.\n\nChange-Id: I64ce824f27838650f350c6138ef497354f49941d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f1a361bd83473a8bea6240bc8257eb9a24f7870b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f1a361bd83473a8bea6240bc8257eb9a24f7870b"}]},"branch":"refs/heads/master"},"d8754a9fc73ef53d1b43fc9c01f85cf009264990":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-12 05:07:37.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/35/675535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/35/675535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/35/675535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/35/675535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/35/675535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/35/675535/2"}}},"commit":{"parents":[{"commit":"97df9e9911cf395abc3c6a6fa7ec525c246fdfb7","subject":"P8: Changes for programming.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/97df9e9911cf395abc3c6a6fa7ec525c246fdfb7"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-09 05:56:36.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-08-12 05:05:52.000000000","tz":-420},"subject":"P8.5: Remove NotImplementedError for some dbapi methods.","message":"P8.5: Remove NotImplementedError for some dbapi methods.\n\nChange-Id: I64ce824f27838650f350c6138ef497354f49941d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8754a9fc73ef53d1b43fc9c01f85cf009264990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8754a9fc73ef53d1b43fc9c01f85cf009264990"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
