)]}'
{"id":"openstack%2Fcyborg~675990","triplet_id":"openstack%2Fcyborg~master~I324bdfc06cdd3a30d5a53a3fe0c5498797b0a3ea","project":"openstack/cyborg","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I324bdfc06cdd3a30d5a53a3fe0c5498797b0a3ea","subject":"Add python 3 unit test job","status":"ABANDONED","created":"2019-08-12 16:58:17.000000000","updated":"2019-09-20 22:24:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11fa4c9f0eadde7b2fb11c9210a5175907e2a686","_number":675990,"virtual_id_number":675990,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":-1,"date":"2019-08-12 17:33:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-12 17:16:22.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":-1,"date":"2019-08-12 17:11:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 17:11:36.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-08-12 17:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5d189260f588b451af3167e2a367653e4ddbfbc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-12 16:58:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77e143d0f67db6b2f1abb6f4ab7204eda4ce6e2e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-12 16:59:45.000000000","message":"Topic set to python3-first","accounts_in_message":[],"_revision_number":1},{"id":"7ce871d13e99e4b29fc48b9b11283ef55ac69190","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-08-12 17:11:36.000000000","message":"Patch Set 1: Code-Review-1\n\nCan we make the Py3 jobs as non-voting?","accounts_in_message":[],"_revision_number":1},{"id":"557458c16516b53ccb618320a6bae76d6fb410b2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-12 17:16:22.000000000","message":"Patch Set 1:\n\n\u003e Can we make the Py3 jobs as non-voting?\n\nI will suggest to add them as voting and fix the failure if any. In the starting of U Cycle, we are planning to remove the py27 support[1] which means every OpenStack project has to run on py3. Which makes Train as a deadline for every projects to be ready for py3.\n\n[1] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html","accounts_in_message":[],"_revision_number":1},{"id":"6782c64d1e741d1e606b3f0f4fc58989c9cc1f32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 17:33:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a293646e71c481fa5d817c957759e0f : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593fbcee426b4d329c9a34a16ad95e81 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c27d9cd2e79143f9ba488b7a35dbda4a : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a97d85a598064c79b85af13e0b8a8239 : FAILURE in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/758ed86c87be4ae4b066ea6f987c2df1 : FAILURE in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c59757d463b42ff914ffec794255af6 : SUCCESS in 4m 01s\n- cyborg-devstack https://zuul.opendev.org/t/openstack/build/2a6acee08f244fa2b1b306fbde5e2ced : SUCCESS in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bacfe00f3a3ac0800a981c6782a0ebf7341534f","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-12 04:29:11.000000000","message":"Patch Set 1:\n\nThis is already included in https://review.opendev.org/667524. We suggest this be abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"11fa4c9f0eadde7b2fb11c9210a5175907e2a686","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-20 22:24:28.000000000","message":"Abandoned\n\nThis is already included in https://review.opendev.org/667524","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9367761fbe2d5f48cfc0c01f6e1939b89d8433b","revisions":{"c9367761fbe2d5f48cfc0c01f6e1939b89d8433b":{"kind":"REWORK","_number":1,"created":"2019-08-12 16:58:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/90/675990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/90/675990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/90/675990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/90/675990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/90/675990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/90/675990/1"}}},"commit":{"parents":[{"commit":"8c26ba469667c1c86d8bf0b31c55cf79f35151b1","subject":"Merge \"Add the fake driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8c26ba469667c1c86d8bf0b31c55cf79f35151b1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-12 16:57:16.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-12 16:58:14.000000000","tz":0},"subject":"Add python 3 unit test job","message":"Add python 3 unit test job\n\nThis commit adds a unit test jobs running\nunder Python 3.6 and python 3.7 as part of\nthe python3-first goal.\n\nSee the python3-first goal document for details:\nhttps://governance.openstack.org/tc/goals/stein/python3-first.html\n\nChange-Id: I324bdfc06cdd3a30d5a53a3fe0c5498797b0a3ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9367761fbe2d5f48cfc0c01f6e1939b89d8433b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9367761fbe2d5f48cfc0c01f6e1939b89d8433b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
