)]}'
{"id":"openstack%2Fcyborg~682820","triplet_id":"openstack%2Fcyborg~master~Iaad80c7aef0f1b2e52600e4632bc395a7d75467b","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"Iaad80c7aef0f1b2e52600e4632bc395a7d75467b","subject":"[wip]api part","status":"ABANDONED","created":"2019-09-18 08:01:12.000000000","updated":"2019-09-18 10:42:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b3468bf6e4bdb2d3ec37773c5efd899e469c879","_number":682820,"virtual_id_number":682820,"owner":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-18 09:13:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 09:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f582568972951b0a0e02aa8a31771c90a2387288","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-09-18 08:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"498bb9915f99fa4ea1234a57b9e77deded93fe03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:13:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4efe047f72045d3a10e31575c456b86 : FAILURE in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab612c81e5234aa9859c0d075ab0a5a3 : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5118e2f2aa340a7935809934248668f : FAILURE in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caf72bdd5d234b37b487193372b9dacd : FAILURE in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ca6a55ecbe4022af5c7e4327607403 : SUCCESS in 5m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d1220408c5a7439f9e6f48e7e2ecaf9c : FAILURE in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b3468bf6e4bdb2d3ec37773c5efd899e469c879","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-09-18 10:42:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14d5f58d38e32ab6dc1b19fc4048898b86fbab13","revisions":{"14d5f58d38e32ab6dc1b19fc4048898b86fbab13":{"kind":"REWORK","_number":1,"created":"2019-09-18 08:01:12.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/20/682820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/20/682820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/20/682820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/20/682820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/20/682820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/20/682820/1"}}},"commit":{"parents":[{"commit":"2530f56b932751ebec6c2540e993c7546944c44d","subject":"[WIP] P6.0: conductor writes device_profile update to db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2530f56b932751ebec6c2540e993c7546944c44d"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-09-18 08:00:57.000000000","tz":-420},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2019-09-18 08:00:57.000000000","tz":-420},"subject":"[wip]api part","message":"[wip]api part\n\nChange-Id: Iaad80c7aef0f1b2e52600e4632bc395a7d75467b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/14d5f58d38e32ab6dc1b19fc4048898b86fbab13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/14d5f58d38e32ab6dc1b19fc4048898b86fbab13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
