)]}'
{"id":"openstack%2Fcyborg~686518","triplet_id":"openstack%2Fcyborg~master~Ia4ca9e161d75bdc27ffc93d76bb82c636056b28e","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"Ia4ca9e161d75bdc27ffc93d76bb82c636056b28e","subject":"API documentation update.","status":"MERGED","created":"2019-10-04 03:15:53.000000000","updated":"2019-10-10 22:04:37.000000000","submitted":"2019-10-10 22:02:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686518-1570744972051-8e81ebfc","meta_rev_id":"4da21cea6d8623ea38085659623535f5b872938e","_number":686518,"virtual_id_number":686518,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 22:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"recommended":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":1,"date":"2019-10-04 18:03:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":2,"date":"2019-10-10 12:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":1,"date":"2019-10-10 13:33:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-10-10 13:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-08 01:23:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":1,"date":"2019-10-10 21:57:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 03:19:29.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2019-10-04 03:19:48.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2019-10-04 03:21:03.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-10-04 18:03:14.000000000","updated_by":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"reviewer":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"state":"REVIEWER"},{"updated":"2019-10-08 01:23:17.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2019-10-10 12:41:53.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-10-10 13:33:27.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-10-10 13:40:41.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-10-10 22:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05babfb5fe9dd5c471319b08c058254e065c22e6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-10-04 03:15:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98b90367727d2bedc04479f38975dedeb2d12bcb","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-10-04 03:23:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77cf8020bd3ce618c736388d5a7683500b0389e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 03:36:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17cfd480e999466f8d8ed4efb6ce0091 : FAILURE in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06100d1d81b94733ace5d3b61d6e3d76 : FAILURE in 12m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d878ee3379bce5ae46466a344673b621e8c0438a","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-10-04 14:06:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20e62f74bec2cd7e3db55c10420de61dacc3ec49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:15:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1a54bdc14348b8b5f66ae8b1991b1d : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef42f6f7050649bb80c049aa18a2b26a : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":3},{"id":"fccce2d4fa197326ab42b16dd0dd3fc436138768","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-10-04 14:23:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"761157845f87b39c41e49da917aa00a177c08850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:34:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553046564ad84e0f942bfd1c9d3f7c1d : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f3dd16b9a8e49389c09f867e3cf8c43 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"69d9272d2cacfb28908ec275ae226eccfc35e377","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-10-04 18:03:14.000000000","message":"Patch Set 4: Code-Review+1\n\nOnly +1 this to show I am happy with it","accounts_in_message":[],"_revision_number":4},{"id":"faf1e4c12c6c2339413613fb3d3bbc8d05090c0e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-10-08 01:23:17.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"4c77355b4e93fa74a2e6690e5f8b2d3c8fefbf52","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-10-10 12:41:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d86557814762870cb3cc44d0c2d5d9416bf8a9f3","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-10 13:33:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4fa3bbcdf8f93177fa62f5994bf736f335aec89c","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-10-10 13:40:41.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"4b2b5f4131616c2832767922b67063744fe599d1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-10-10 21:57:18.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8ba10180e16660e3fab6d398e7957a34a56d12b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 21:57:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a03a50f4698b98c34ca1b8c63fc941c6c96909eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 22:02:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531d74595d7541b9b8c1d14a5edf97a5 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b708e1ba768458a9d0a337fd7418a0d : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"8d8ba0de34df5775ca0563fbfcc9c1effbd06d9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 22:02:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4da21cea6d8623ea38085659623535f5b872938e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 22:04:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4588c1fff5404c5c9772250db59c605b : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52494d89f700e6c68bc547180f6cf2a361740b02","revisions":{"c7b9b8b68fdae3f3f42d5f4882d850ecfde6f9a9":{"kind":"REWORK","_number":1,"created":"2019-10-04 03:15:53.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/18/686518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/686518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/686518/1"}}},"commit":{"parents":[{"commit":"cc70978dd04b3866aa7db3b3889b91b539e3e725","subject":"Update driver section.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cc70978dd04b3866aa7db3b3889b91b539e3e725"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-10-04 03:14:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-10-04 03:14:21.000000000","tz":-420},"subject":"API documentation update.","message":"API documentation update.\n\nChange-Id: Ia4ca9e161d75bdc27ffc93d76bb82c636056b28e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c7b9b8b68fdae3f3f42d5f4882d850ecfde6f9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c7b9b8b68fdae3f3f42d5f4882d850ecfde6f9a9"}]},"branch":"refs/heads/master"},"11dd8edbebbd7c10696739704a57b4a6658dbd93":{"kind":"REWORK","_number":2,"created":"2019-10-04 03:23:26.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/18/686518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/686518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/686518/2"}}},"commit":{"parents":[{"commit":"cc70978dd04b3866aa7db3b3889b91b539e3e725","subject":"Update driver section.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cc70978dd04b3866aa7db3b3889b91b539e3e725"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-10-04 03:14:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-10-04 03:22:24.000000000","tz":-420},"subject":"API documentation update.","message":"API documentation update.\n\nChange-Id: Ia4ca9e161d75bdc27ffc93d76bb82c636056b28e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/11dd8edbebbd7c10696739704a57b4a6658dbd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/11dd8edbebbd7c10696739704a57b4a6658dbd93"}]},"branch":"refs/heads/master"},"c1be29fd771f0fbb72bc19fa06e4e04870bdf896":{"kind":"REWORK","_number":3,"created":"2019-10-04 14:06:14.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/18/686518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/686518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/686518/3"}}},"commit":{"parents":[{"commit":"cc70978dd04b3866aa7db3b3889b91b539e3e725","subject":"Update driver section.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cc70978dd04b3866aa7db3b3889b91b539e3e725"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-10-04 03:14:21.000000000","tz":-420},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-10-04 13:07:19.000000000","tz":-240},"subject":"API documentation update.","message":"API documentation update.\n\nChange-Id: Ia4ca9e161d75bdc27ffc93d76bb82c636056b28e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c1be29fd771f0fbb72bc19fa06e4e04870bdf896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c1be29fd771f0fbb72bc19fa06e4e04870bdf896"}]},"branch":"refs/heads/master"},"52494d89f700e6c68bc547180f6cf2a361740b02":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-04 14:23:25.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/18/686518/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/18/686518/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/18/686518/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/18/686518/4"}}},"commit":{"parents":[{"commit":"95e01e6d38ff6ab5b4c22a780e7dbab9784d05de","subject":"Update driver section.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/95e01e6d38ff6ab5b4c22a780e7dbab9784d05de"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-10-04 03:14:21.000000000","tz":-420},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-10-04 14:23:25.000000000","tz":0},"subject":"API documentation update.","message":"API documentation update.\n\nChange-Id: Ia4ca9e161d75bdc27ffc93d76bb82c636056b28e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/52494d89f700e6c68bc547180f6cf2a361740b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/52494d89f700e6c68bc547180f6cf2a361740b02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
