)]}'
{"id":"openstack%2Fcyborg~699906","triplet_id":"openstack%2Fcyborg~master~Id9b6656b14298f1bfff01f5a195638e516109907","project":"openstack/cyborg","branch":"master","topic":"fix_get_device_id","hashtags":[],"change_id":"Id9b6656b14298f1bfff01f5a195638e516109907","subject":"Improve objects/device.py UT coverage from 82% to 100%","status":"MERGED","created":"2019-12-19 02:53:12.000000000","updated":"2019-12-31 06:34:31.000000000","submitted":"2019-12-31 06:33:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699906-1577773985889-a23e576b","meta_rev_id":"3dc378d11fbda3a484f86110b02ab2dd6c237c50","_number":699906,"virtual_id_number":699906,"owner":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-31 06:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-31 03:41:26.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-12-30 07:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2019-12-31 05:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2019-12-31 05:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 02:55:38.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-12-19 02:55:44.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-12-19 02:55:58.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2019-12-23 04:13:29.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-12-30 07:24:30.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-12-31 05:58:26.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-12-31 06:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6971ad3ff6931b75b049599920e72a3e301b153","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-19 02:53:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a150659812299f46cb3d34506a98bcfda64568d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 03:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/219a7f8f33b94d8789037d972f209f06 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9e4e674b304d22b868aced7b3aa732 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2cd0f669736460f9f8821a5e88e9690 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0dcb10d04b4a49fe9e2670a73b5c3a1b : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a56daf7499d4089903f5fd6ad7d3e47 : SUCCESS in 4m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c94be654b9754ab2b8e97e203bd8f18e : SUCCESS in 29m 53s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86a4c3a0351e4afbbbe8d1e5831cc106 : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71c78cd90f0873a84e74a88a7144e8806b2180d9","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-19 11:28:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e91b2b07151ae3b8687d2da17d38d588e765de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 12:21:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff63fb21056649778cf114b10b15cc1b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a9863ca36c439d974a94f846778c06 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/162762973dc04961b5a9baa74dde3f2c : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5081eef5a2794d708986121c2479b468 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad76e8a86d048b4bdb63c68da66a4c9 : SUCCESS in 4m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/971859d599834a9eb345d832417bbf98 : SUCCESS in 47m 05s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7d4fe150ff64f27ae7e62bc166f3dfa : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bef7bb729a3279b9221b548ba6995ca64b020fc","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-12-20 03:58:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nSoft -1.","accounts_in_message":[],"_revision_number":2},{"id":"d9925faaff78c1cbda22808cf32da11ffcbbeece","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-21 02:32:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b9ccc12b57782daebd22fb9926359845f934e24","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-21 02:47:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eca5358042fff57adb399d99a9cd00daaca6a185","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-21 02:47:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f43cdf169bc8ced2040ab46a194d1acace46d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 03:06:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3a5fa25412c44cab34318543e48a326 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b242af9d42b3402b8715c15c95a3840d : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cecb5d0d62b43f3acb26332bd4a0739 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c484a37ba0e4776a209f59c67a47708 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d847ad6121d4c99bd2b4fac945b1eb0 : SUCCESS in 4m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b042f0fa3e874b9d941ca30c38453bae : SUCCESS in 31m 05s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0b9ec8061eb49b386d4fdade70179ae : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e5a19843543b238fd02fea930ed99ef7af1e9def","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-12-21 17:13:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\nWould like more review.","accounts_in_message":[],"_revision_number":3},{"id":"8d0fb6b9493347830f65a8b2ff6610561a0e8550","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-23 02:40:47.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1f64b67b7a0766f879b3992490c53b7d64ff2aaf","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-23 02:49:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3be0a8f6b32e4d8a085b78d61d263b47dced5027","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-23 02:50:36.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"aa1f1787ad37251aca75e533537493abc5715d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 03:24:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0136d982dd744a8864a7112576e94a6 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5beed66d92cf487bb0c1b6e17a435d68 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d413f4c48d0b4e3bb196f98f11fd35bd : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da3ec1620e9e4d56b3ddea9f852c8e60 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafd7605d1bf4dcf92ee1c41043d18ae : SUCCESS in 3m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/feb1247ea1d147caa9f2451b8d138954 : SUCCESS in 31m 29s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a84f12a2640454ebbd575d28d21c49a : SUCCESS in 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"143a829367f17d852a0f9e67124096b64c2be0d1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-12-23 04:13:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a86e8aa8cdb9bdd91569146e5386dabd92b0f07e","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-12-30 02:17:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nHi chenker, pls see my comments inline.","accounts_in_message":[],"_revision_number":4},{"id":"4fdccf6f377375443b8cbe5ccc12ffedd67f851c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-30 02:32:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8c9cb20c937a23a4f89e0ddecc794a50adcde59","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-30 02:41:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"126e2f723b6f0a82ab04316414b08b94f9d40ed8","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-30 02:42:39.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"024b1bd6fa1cb5db88fed63baa4d77d59a37633d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 03:18:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11d8fd48fa1b449e9d9fa2b6698a1d2c : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf0dd1e201a416898af88b8ea1db48c : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261dffe8648342cea931f72ebc89d34f : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b37f0711b0ae4fba8cd1515433fcc4a5 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1a10dbd0ad41919e8506e1c60e0eb7 : SUCCESS in 3m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d3bfb626943246feb0c5a2732fff3b55 : SUCCESS in 29m 02s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8a1423bdb6a4aacaabc6e48b1ffe893 : SUCCESS in 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4801fd8764ac024f7261d747c17220f5e9a7bd7d","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-12-30 07:24:30.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"eb4b5de47d4373cf8796c8e92b83c53ccdf3bd5f","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-31 02:25:11.000000000","message":"Patch Set 5:\n\nLGTM overall, one suggestion here: maybe it is better to get mutiple fake devices from get_test_devices and test the filter functions?","accounts_in_message":[],"_revision_number":5},{"id":"48b7ef5281099ebea8b1c4b8a0669fad3b670dc2","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-31 03:31:54.000000000","message":"Patch Set 5:\n\n\u003e LGTM overall, one suggestion here: maybe it is better to get\n \u003e mutiple fake devices from get_test_devices and test the filter\n \u003e functions?\n\nThanks Xinran. The UT for test_list_with_filter is to test list method the situation using filters.  The important things to make the  method list logic self-consistent . We do not need to pay much attention to the return value of the mock method.","accounts_in_message":[],"_revision_number":5},{"id":"8d55ba099f1b4d095e04bb4377e68ae8ae0d6781","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-31 03:35:46.000000000","message":"Patch Set 5:\n\n\u003e \u003e LGTM overall, one suggestion here: maybe it is better to get\n \u003e \u003e mutiple fake devices from get_test_devices and test the filter\n \u003e \u003e functions?\n \u003e \n \u003e Thanks Xinran. The UT for test_list_with_filter is to test list\n \u003e method the situation using filters.  The important things to make\n \u003e the  method list logic self-consistent . We do not need to pay much\n \u003e attention to the return value of the mock method.\n\nIf we have only one fake device, how do we know the limit\u003d1 filter succeed? Also we can have 2 devices from different host, and we can test the get_by_hostname then. It is nice to have, just a suggestion :)","accounts_in_message":[],"_revision_number":5},{"id":"148906245dbc5f30f3000995c0c63c75f6e3971d","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-31 03:41:26.000000000","message":"Patch Set 5:\n\n\u003e \u003e \u003e LGTM overall, one suggestion here: maybe it is better to get\n \u003e \u003e \u003e mutiple fake devices from get_test_devices and test the filter\n \u003e \u003e \u003e functions?\n \u003e \u003e\n \u003e \u003e Thanks Xinran. The UT for test_list_with_filter is to test list\n \u003e \u003e method the situation using filters.  The important things to make\n \u003e \u003e the  method list logic self-consistent . We do not need to pay\n \u003e much\n \u003e \u003e attention to the return value of the mock method.\n \u003e \n \u003e If we have only one fake device, how do we know the limit\u003d1 filter\n \u003e succeed? Also we can have 2 devices from different host, and we can\n \u003e test the get_by_hostname then. It is nice to have, just a\n \u003e suggestion :)\n\nYe. At least two device then we can vifity whether the limit is good or not. But this method is to test https://github.com/openstack/cyborg/blob/master/cyborg/objects/device.py#L62. If we want to use a single method to test cls.dbapi.device_list_by_filters in https://github.com/openstack/cyborg/blob/master/cyborg/objects/device.py#L69 or Function test. We should take your suggestion. Anyway, your suggestion is good. We should pay attention later.","accounts_in_message":[],"_revision_number":5},{"id":"b4fb76be3d8153b7ee185442ee0514e5f783e3de","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-31 05:58:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n\u003e \u003e \u003e \u003e LGTM overall, one suggestion here: maybe it is better to get\n \u003e \u003e \u003e \u003e mutiple fake devices from get_test_devices and test the\n \u003e filter\n \u003e \u003e \u003e \u003e functions?\n \u003e \u003e \u003e\n \u003e \u003e \u003e Thanks Xinran. The UT for test_list_with_filter is to test list\n \u003e \u003e \u003e method the situation using filters.  The important things to\n \u003e make\n \u003e \u003e \u003e the  method list logic self-consistent . We do not need to pay\n \u003e \u003e much\n \u003e \u003e \u003e attention to the return value of the mock method.\n \u003e \u003e\n \u003e \u003e If we have only one fake device, how do we know the limit\u003d1\n \u003e filter\n \u003e \u003e succeed? Also we can have 2 devices from different host, and we\n \u003e can\n \u003e \u003e test the get_by_hostname then. It is nice to have, just a\n \u003e \u003e suggestion :)\n \u003e \n \u003e Ye. At least two device then we can vifity whether the limit is\n \u003e good or not. But this method is to test https://github.com/openstack/cyborg/blob/master/cyborg/objects/device.py#L62.\n \u003e If we want to use a single method to test cls.dbapi.device_list_by_filters\n \u003e in https://github.com/openstack/cyborg/blob/master/cyborg/objects/device.py#L69\n \u003e or Function test. We should take your suggestion. Anyway, your\n \u003e suggestion is good. We should pay attention later.\n\nSure, we can improve later.","accounts_in_message":[],"_revision_number":5},{"id":"5d6c1d566e7f4bbf2205856eea7e8030c0abba46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 05:58:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e0410efac5150926c96a0eef96c9edecc7f28776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 06:33:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73db06872f1e49cfbf313abc033ece32 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86a56dc34b5c49b39d121b1f49adce68 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9dc72a4bc3a4db190886030a8fdee20 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9bc73b3aac4a37841ec9e9d78ebc87 : SUCCESS in 5m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/641b6aad52604148922df867e7998f7b : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":5},{"id":"8dfef8ff71413ae4e53ea16153b02c885c74f5c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 06:33:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e7ce64d089254d9e212a98d5c17126bd6b103acc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 06:34:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dca48746e71490686af2368a1106e10 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7d98ff25993d5dd2acd3c88650fe39aa66688bff","revisions":{"63627140ae05b25727028c07959320ab86f7be69":{"kind":"REWORK","_number":1,"created":"2019-12-19 02:53:12.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/06/699906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/06/699906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/06/699906/1"}}},"commit":{"parents":[{"commit":"59b8047ea15793ca0b71acce8cdc35533eea3d7a","subject":"Refactor _gen_resource_inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/59b8047ea15793ca0b71acce8cdc35533eea3d7a"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 02:49:36.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 02:49:36.000000000","tz":480},"subject":"Improve objects/device.py UT coverage from 76% to 100%","message":"Improve objects/device.py UT coverage from 76% to 100%\n\nChange-Id: Id9b6656b14298f1bfff01f5a195638e516109907\nStory: 2007036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63627140ae05b25727028c07959320ab86f7be69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63627140ae05b25727028c07959320ab86f7be69"}]},"branch":"refs/heads/master"},"ba3cee1ec72da48d83a1c8e13f9afc1d61b991bb":{"kind":"REWORK","_number":2,"created":"2019-12-19 11:28:24.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/06/699906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/06/699906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/06/699906/2"}}},"commit":{"parents":[{"commit":"5a959d854dcb70bcdc3aed0a0b23a9b07efa94a2","subject":"BugFix: Modify objects/device.py get_by_device_id method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5a959d854dcb70bcdc3aed0a0b23a9b07efa94a2"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 02:49:36.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 11:26:40.000000000","tz":480},"subject":"Improve objects/device.py UT coverage from 82% to 100%","message":"Improve objects/device.py UT coverage from 82% to 100%\n\nChange-Id: Id9b6656b14298f1bfff01f5a195638e516109907\nStory: 2007036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ba3cee1ec72da48d83a1c8e13f9afc1d61b991bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ba3cee1ec72da48d83a1c8e13f9afc1d61b991bb"}]},"branch":"refs/heads/master"},"ecd9bc527518298a3cba9512a807cd70bef33acf":{"kind":"REWORK","_number":3,"created":"2019-12-21 02:32:11.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/06/699906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/06/699906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/06/699906/3"}}},"commit":{"parents":[{"commit":"b35f6ea72b49dcdb99f0748bdb003b872afe1ba7","subject":"BugFix: Modify objects/device.py get_by_device_id method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b35f6ea72b49dcdb99f0748bdb003b872afe1ba7"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 02:49:36.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-20 10:16:39.000000000","tz":480},"subject":"Improve objects/device.py UT coverage from 82% to 100%","message":"Improve objects/device.py UT coverage from 82% to 100%\n\nChange-Id: Id9b6656b14298f1bfff01f5a195638e516109907\nStory: 2007036\nTask: 37842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ecd9bc527518298a3cba9512a807cd70bef33acf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ecd9bc527518298a3cba9512a807cd70bef33acf"}]},"branch":"refs/heads/master"},"f9e45edca30d804dd371e8cc05883386c1508033":{"kind":"REWORK","_number":4,"created":"2019-12-23 02:49:50.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/06/699906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/06/699906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/06/699906/4"}}},"commit":{"parents":[{"commit":"b35f6ea72b49dcdb99f0748bdb003b872afe1ba7","subject":"BugFix: Modify objects/device.py get_by_device_id method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b35f6ea72b49dcdb99f0748bdb003b872afe1ba7"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 02:49:36.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-23 02:49:14.000000000","tz":480},"subject":"Improve objects/device.py UT coverage from 82% to 100%","message":"Improve objects/device.py UT coverage from 82% to 100%\n\nChange-Id: Id9b6656b14298f1bfff01f5a195638e516109907\nStory: 2007036\nTask: 37842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f9e45edca30d804dd371e8cc05883386c1508033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f9e45edca30d804dd371e8cc05883386c1508033"}]},"branch":"refs/heads/master"},"7d98ff25993d5dd2acd3c88650fe39aa66688bff":{"kind":"REWORK","_number":5,"created":"2019-12-30 02:41:51.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/06/699906/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/06/699906/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/699906/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/06/699906/5"}}},"commit":{"parents":[{"commit":"b35f6ea72b49dcdb99f0748bdb003b872afe1ba7","subject":"BugFix: Modify objects/device.py get_by_device_id method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b35f6ea72b49dcdb99f0748bdb003b872afe1ba7"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-19 02:49:36.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-12-30 02:34:42.000000000","tz":480},"subject":"Improve objects/device.py UT coverage from 82% to 100%","message":"Improve objects/device.py UT coverage from 82% to 100%\n\nChange-Id: Id9b6656b14298f1bfff01f5a195638e516109907\nStory: 2007036\nTask: 37842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7d98ff25993d5dd2acd3c88650fe39aa66688bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7d98ff25993d5dd2acd3c88650fe39aa66688bff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
