)]}'
{"id":"openstack%2Fcyborg~702578","triplet_id":"openstack%2Fcyborg~master~Ia6f415b74150cdda94d71dc8b5e760f7758a526e","project":"openstack/cyborg","branch":"master","topic":"improve_dp_db_ut","hashtags":[],"change_id":"Ia6f415b74150cdda94d71dc8b5e760f7758a526e","subject":"Improve UT for cyborg/db device profile","status":"MERGED","created":"2020-01-15 03:28:47.000000000","updated":"2020-02-08 17:44:09.000000000","submitted":"2020-02-08 17:41:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702578-1581183701352-bbef5f2b","meta_rev_id":"944ed6c955eebe85fcc746dffec0df78289a8f5e","_number":702578,"virtual_id_number":702578,"owner":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-08 17:41:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":2,"date":"2020-01-19 16:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2020-01-20 09:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":1,"date":"2020-01-20 10:03:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2020-01-20 09:54:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 03:35:14.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2020-01-15 03:35:19.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-01-15 03:35:23.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-01-15 03:35:29.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-01-15 03:35:34.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2020-01-19 16:44:03.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-01-20 09:54:32.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-01-20 10:03:44.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-08 17:41:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3382cb2c1ec745892a2387b75e678d17df470f0e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-15 03:28:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdfc060aa8e77461f9ee2e1957cff92e0a96efe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 04:00:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d59a1c2ed5c49d6ab09207bc55adfc3 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973a83affd3b4c8d845ee27ce8db7eab : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf8058a6f624b63b227b5410c1761eb : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e65776e7e2c943dcb1cdf6df5746d6a3 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77cb9fe146c6462784d6b0b4fdfb55d5 : SUCCESS in 4m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/44afd2a01dde40d7bad7865a65992691 : SUCCESS in 26m 53s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbaf0c2a16ef4e7d91fc6a17d23b1d5d : SUCCESS in 30m 20s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/871b0d2b44f74a4e8dd825da7cdce277 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e435273c4031d393268c8c81a052a6926db364c9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-15 06:22:02.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nDoes we need to remove the \"context\" parameter from device_profile create api and rpcapi ?","accounts_in_message":[],"_revision_number":1},{"id":"08589b5eecfa55db11845c03e74a1f4246e5a842","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-15 07:04:53.000000000","message":"Patch Set 1: Code-Review-1\n\nSoft -1, just as a reminder.\n\nI see that device_profile post api [1] passed \"context\" parameter, and conductor rpcapi [2] received this parameter. Now if you remove the \"context\" parameter, do you want to remove the above parameters as well? We did not use this parameter after all.\n\nBut I\u0027m not sure if rpcapi must have this parameter, as IRC communication, which will be discussed on January 16, 2020 (the regular meeting on Thursday at 11:00 （UTC 3:00）).\n\n[1]https://github.com/openstack/cyborg/blob/d8bd76ea8d54e2bf9a89ab4fa126a81c6edd7c78/cyborg/api/controllers/v2/device_profiles.py#L116-L117\n\n[2]https://github.com/openstack/cyborg/blob/d8bd76ea8d54e2bf9a89ab4fa126a81c6edd7c78/cyborg/conductor/rpcapi.py#L156","accounts_in_message":[],"_revision_number":1},{"id":"6a1982ea44fb665a9f92c535957f017bf3b27430","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-15 08:03:54.000000000","message":"Patch Set 1:\n\n\u003e Soft -1, just as a reminder.\n \u003e \n \u003e I see that device_profile post api [1] passed \"context\" parameter,\n \u003e and conductor rpcapi [2] received this parameter. Now if you remove\n \u003e the \"context\" parameter, do you want to remove the above parameters\n \u003e as well? We did not use this parameter after all.\n \u003e \n \u003e But I\u0027m not sure if rpcapi must have this parameter, as IRC\n \u003e communication, which will be discussed on January 16, 2020 (the\n \u003e regular meeting on Thursday at 11:00 （UTC 3:00）).\n \u003e \n \u003e [1]https://github.com/openstack/cyborg/blob/d8bd76ea8d54e2bf9a89ab4fa126a81c6edd7c78/cyborg/api/controllers/v2/device_profiles.py#L116-L117\n \u003e \n \u003e [2]https://github.com/openstack/cyborg/blob/d8bd76ea8d54e2bf9a89ab4fa126a81c6edd7c78/cyborg/conductor/rpcapi.py#L156\n\nThanks for your review. I read the rpc code[1]. The context is necessary. And I try to remove context in my devstack env. It got error \"call() takes exactly 3 arguments (2 given)\". So the conclusion from test is that we can\u0027t remove this before rpc call.","accounts_in_message":[],"_revision_number":1},{"id":"61166d987afd341cc020d57c30b4d2492702514d","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-15 08:04:33.000000000","message":"Patch Set 1:\n\nrpc code[1].\n[1]. https://github.com/openstack/oslo.messaging/blob/e7420afa2f8eb42d17f9af2ad90ed0090a89ed41/oslo_messaging/rpc/client.py#L158","accounts_in_message":[],"_revision_number":1},{"id":"9c30d9cbf2cbbfd598784a7d47c13e98b1f66689","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-15 08:10:59.000000000","message":"Patch Set 1:\n\n\u003e \n \u003e Thanks for your review. I read the rpc code[1]. The context is\n \u003e necessary. And I try to remove context in my devstack env. It got\n \u003e error \"call() takes exactly 3 arguments (2 given)\". So the\n \u003e conclusion from test is that we can\u0027t remove this before rpc call.\n\nThanks, if so, I think we need to consider the consistency of the interface parameters or modify them in the current way (just remove the context from the sqlarmy api).","accounts_in_message":[],"_revision_number":1},{"id":"5e3e8575c92f72e635f8df405f1b99c053096763","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-15 08:30:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53ee65b96d9f172ccd757da5798b399f1a9990c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 09:11:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/789cda9897694ddd9589b61652941608 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfc2368d5344b348e5f1841536511b3 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053692408c534e90bef538b32a24fa12 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59b277f536414d0abbec47d2593f6447 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d21db37ae764f83a42fb161b35a8e94 : SUCCESS in 3m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/73297667bea84dbda42d008cccaddb80 : SUCCESS in 29m 06s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9dcb0dcb0f74914bab476e7bac56edc : SUCCESS in 35m 53s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/7a55906b29f44521baab963f41f9fc80 : FAILURE in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd7820c2bed9abc6d80afee7bb79b4ae69fb3759","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-16 07:25:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a0b66ab781d76e18c7183a70ed753046a4bd3281","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-16 07:29:26.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c901d06fda7981a7c59ba55c0f96e81dbf9c59f4","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-16 08:06:47.000000000","message":"Patch Set 4:\n\n@Sundar. Hi, What is your opinion？","accounts_in_message":[],"_revision_number":4},{"id":"c2a3d2b05c372866a40860a4f772cc7dcd28d8a4","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-16 08:14:35.000000000","message":"Patch Set 4:\n\nI refer to the code implementation of watcher, it just does not pass in useless parameters.\n[1]. https://github.com/openstack/watcher/blob/master/watcher/tests/db/utils.py#L226\n[2]. https://github.com/openstack/watcher/blob/master/watcher/db/sqlalchemy/api.py#L481\nI tried to remove the context as a whole, but found that the rpc[3][4] must have the context. Therefore, I think it is reasonable to delete only the relevant part of the database.\n[3]. https://github.com/openstack/oslo.messaging/blob/e7420afa2f8eb42d17f9af2ad90ed0090a89ed41/oslo_messaging/rpc/client.py#L158\n[4]. rpc error when removing context param: \"call() takes exactly 3 arguments (2 given)\"","accounts_in_message":[],"_revision_number":4},{"id":"fc9dc3134b669078faf0cb516483f1b567d3b522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:23:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e3b164ea3cb4550a0bafdea7842cdb2 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670c9f64b2b14a18a048c6a3d27ae68d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b6f4dda55f749098626629e7b526391 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb05b03971304bc5a1ac41e8974b049c : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7072ca420fef4b3fbe27661cf064b9af : SUCCESS in 5m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ca42effd1301403bbe2979094e5410b3 : SUCCESS in 43m 39s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e72ffb04673343f29a341b326e0b0999 : SUCCESS in 49m 45s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/a0644c775e80495492c3ff39482b3746 : FAILURE in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cadd1983bfdf9d30c0a214d0f6fdd60a44f4d944","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-16 12:24:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6305dbffb7bac8b9b590270323576bc439556088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 13:07:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f40162e0347a4ea7a02d16e9b893ccee : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928c532544a14e5fa78f805c7e6f544f : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbe9141e5f9e4b4fb179f8192d6e5546 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/723d2f9fa44848f9a7458f54546ff70c : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81afc70e96a04e349b976a6014132893 : SUCCESS in 5m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5ab148e4ada84320a03b26324adb9a92 : SUCCESS in 38m 33s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ae11d3ad3c044e09f4c81d104ec4736 : SUCCESS in 33m 01s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/153445548307481a8a2915bd3fff1ccb : FAILURE in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e933bd8afe0f862fbc97ec7bb267a8645e4cbd9","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-17 05:12:33.000000000","message":"Patch Set 5: Code-Review-1\n\nSoft -1 till we answer this question below.\n\nFor the context parameter in device_profile_create, I have a basic question. Without the context parameter, we have only the RBAC role [1] to enforce user/tenant permissions. Is that enough? If we need to enforce per-project permissions, quotas, etc., would that not need the context at least till the objects layer?\n\n[1] https://opendev.org/openstack/cyborg/src/branch/master/cyborg/api/controllers/v2/device_profiles.py#L90.","accounts_in_message":[],"_revision_number":5},{"id":"f1f0dccae08bdc7a86cd57fe22943acea7d61834","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 06:36:30.000000000","message":"Patch Set 5:\n\n\u003e Soft -1 till we answer this question below.\n \u003e \n \u003e For the context parameter in device_profile_create, I have a basic\n \u003e question. Without the context parameter, we have only the RBAC role\n \u003e [1] to enforce user/tenant permissions. Is that enough? If we need\n \u003e to enforce per-project permissions, quotas, etc., would that not\n \u003e need the context at least till the objects layer?\n \u003e \n \u003e [1] https://opendev.org/openstack/cyborg/src/branch/master/cyborg/api/controllers/v2/device_profiles.py#L90.\n\nIf we consider the restrictions on quotas for different projects, it is possible to need context in the db layer. So don\u0027t delete it.","accounts_in_message":[],"_revision_number":5},{"id":"b59da178fae2b0fd085c49b4608953cd4f90a711","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 07:37:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a44a8ebb29833fc89b4d8ccb731616c822a69f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 08:17:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b35c2b734614477398f33a2689c926ec : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5728b19e5794fe7a9c8a26283258fe9 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee58197a24324d27a4099153a9e49559 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01f5880db2124e83a75b0441240376b7 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439c3415f781492687864f9b62af06ea : SUCCESS in 5m 07s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e18fcf2381af426187101c03f4ab9cc3 : SUCCESS in 29m 42s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/742bb2d4dd0e4429a627ea0eb920eaaf : SUCCESS in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"53457931a17e7f112eba1f24a413d4ee26943c3a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-17 08:18:05.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5f6cc849c7f11a4ece4b3bae596341f288dfa934","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 08:46:44.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"37600fc75fa865555894a43604629ca2865aceee","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 08:51:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"61f8381f81f04b78bde824be54fc39870602c984","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 08:54:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7fb6d27fbbf576b8bcae556811fb0f41af9500b9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-17 09:11:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3aa8fe8cadbb469512f2a40b6e8ec8c4a828b81a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-17 09:23:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ba3963fc01800b79e988da651e874c9d16cdc067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 09:36:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/085ee7ce9c3042d78cb138ecb4e91dc3 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09543aef6d948278484f9c0cc558376 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cd87eebd4c742b9a8a2f6c3f45f3b64 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7050b98d288a4882af5482777c963a6a : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139783d4e2f1448eaa7e454814f88f77 : SUCCESS in 5m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/278a137ff9bf4b308eec1a8237280e27 : SUCCESS in 34m 16s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68629bbf8b3e417c9655f8032f1243a9 : SUCCESS in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"325eddf391c9bbc92e541ae5aac393b860dd27f9","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 09:38:23.000000000","message":"Patch Set 8:\n\nBy the error message, we can konw that we should call it by filters \u003d {} If we want to test filters param.  Because None can\u0027t be iterator.","accounts_in_message":[],"_revision_number":8},{"id":"6c33ef6268766d9dde50b7cbba22d13d79e4b30a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-17 09:58:06.000000000","message":"Patch Set 8:\n\n(1 comment)\n\n\u003e By the error message, we can konw that we should call it by filters\n \u003e \u003d {} If we want to test filters param.  Because None can\u0027t be\n \u003e iterator.\n\nYeah, as talked on IRC, I was do a follow up, this patch LGTM.","accounts_in_message":[],"_revision_number":8},{"id":"33e69b07053fe3d63b5fd90b4355be6ab836c487","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-17 09:58:12.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"6214e975cdc3b18c3f0c645e4aa1ecc76a6236e4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-18 17:33:06.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nSoft -1","accounts_in_message":[],"_revision_number":8},{"id":"9b04052e933ccc81cd62eed213dd98fabae8ecd6","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 01:35:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cac409b56d0f7f1beb65c1191bd060bdac5abf3a","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 01:40:10.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nThanks.","accounts_in_message":[],"_revision_number":9},{"id":"b8de3ad9233719db8f50be8df2f057f51651bc9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 02:18:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9eadba5508e4ce8a1d192f2f337645d : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2d2e066d32470a88a8cc085b526f29 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fd9892d59b6448fa16ae09eac1f4a8d : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6896b1de4e92412b8329dbc4c7af22de : TIMED_OUT in 41m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876479bf74674fa3a21b34d0bf18a28d : SUCCESS in 3m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/18ed9b2419984ef0b853f9b5c6d73a20 : SUCCESS in 27m 46s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f68b469d36f6404a96e1f30ce233e779 : SUCCESS in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2b7153970d92a154ce6eb2491fa998006a147815","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 03:41:42.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"6db7d2f9a4ba71855443cf0546a29b4b4b32a512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 04:21:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1b709db69d440c281075612e360081a : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3267fddf52d649c580222daa39fe824a : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab72e7cf19b47109bce21d433ef2c61 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c05354e01558471cb01b5b0f7762a09a : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47000e32521b4b9eaf11217e48c76133 : SUCCESS in 3m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/36d753e1cbb1401f8b8e892736fb5f90 : SUCCESS in 26m 47s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14b33e9861a3474a94682f9c48def5c1 : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2b2cc6c6dfc1719e2f928c8242bd722ab020147a","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 07:32:18.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"ce12f27d5ed182fa4e9606c88a7997ea6c4a4d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 08:06:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55f579fe053c444688baab88c723f415 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668d65c98f1941b088ed170f7296bdca : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f23ab5bbeea4492dab4318b32197c87c : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fb0f51fbea34c58aaf48df2575bd0a5 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe3cbc6237d44959aa202766c0c7355 : SUCCESS in 4m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/03f67d9729144b7ca68ac95ca4b1c673 : SUCCESS in 26m 41s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6aba14f3d9994fca9198f9a7eec5e9c8 : SUCCESS in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f3cbee4f861970bec90ad684265ddee6c8599cb2","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-19 16:44:03.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"cd209b0b370d6af222a2c0bdbfcedb2f52f1c497","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-20 09:54:32.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"0211e07d99740a808f5bdc810996c5cd43f63f7b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-20 10:03:44.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"d884d49f043710f546bcfb55e0ce1fc7a20f33c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:07:24.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"995ed22f477de667c4dca2c10caaa94493b02039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:41:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"56a341decbb9f2840efdfbf311f2175260b9634a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:41:41.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed054a235e1c4f688e6a133b459d081b : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bf13bc7637a453981d5db51b8908337 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b367770255149cb9a88b3e863097437 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec9498f31004a4483b8132b1a02846e : SUCCESS in 3m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d11b0f3b068046b0bd495af1a320e057 : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":10},{"id":"6c365410de5bef04b082c44c5e5a882f90f80e4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:44:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e19ecc7124419890151b5f34c0390a : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"6b28ca2722885f4eb4e0ae40b26b19d2347e21c0","revisions":{"ed83ebeda9c594e8807245b3b1104382a6f7f1ba":{"kind":"REWORK","_number":1,"created":"2020-01-15 03:28:47.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/1"}}},"commit":{"parents":[{"commit":"47ebab753cb52e56082d7d11b73e754387f893db","subject":"Merge \"Add created_time and update_time for device profile API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/47ebab753cb52e56082d7d11b73e754387f893db"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-15 03:27:17.000000000","tz":480},"subject":"Improve UT for DB device profile","message":"Improve UT for DB device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ed83ebeda9c594e8807245b3b1104382a6f7f1ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ed83ebeda9c594e8807245b3b1104382a6f7f1ba"}]},"branch":"refs/heads/master"},"c88300113c76a74d866f64e59bc85c1436fcd709":{"kind":"REWORK","_number":2,"created":"2020-01-15 08:30:41.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/2"}}},"commit":{"parents":[{"commit":"47ebab753cb52e56082d7d11b73e754387f893db","subject":"Merge \"Add created_time and update_time for device profile API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/47ebab753cb52e56082d7d11b73e754387f893db"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-15 08:29:15.000000000","tz":480},"subject":"Improve UT for DB device profile","message":"Improve UT for DB device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nAlso, This patch remove unused param context when\ncreating device profile.\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c88300113c76a74d866f64e59bc85c1436fcd709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c88300113c76a74d866f64e59bc85c1436fcd709"}]},"branch":"refs/heads/master"},"ac583dda06ad341e4ab8eb616f5c7ee1128ce47a":{"kind":"REWORK","_number":3,"created":"2020-01-16 07:25:42.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/3"}}},"commit":{"parents":[{"commit":"bad6c9c6df1a00034d27a6c4310a15ede0d293b4","subject":"P1: public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bad6c9c6df1a00034d27a6c4310a15ede0d293b4"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 07:24:46.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nAlso, This patch remove unused param context when\ncreating device profile.\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ac583dda06ad341e4ab8eb616f5c7ee1128ce47a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ac583dda06ad341e4ab8eb616f5c7ee1128ce47a"}]},"branch":"refs/heads/master"},"a43428fc9afcc4f8db2b0d67134cd1068c910068":{"kind":"NO_CHANGE","_number":4,"created":"2020-01-16 07:29:26.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/4"}}},"commit":{"parents":[{"commit":"2a0a0939468adccf86ee0495d271f2608d04806c","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2a0a0939468adccf86ee0495d271f2608d04806c"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 07:28:39.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nAlso, This patch remove unused param context when\ncreating device profile.\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a43428fc9afcc4f8db2b0d67134cd1068c910068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a43428fc9afcc4f8db2b0d67134cd1068c910068"}]},"branch":"refs/heads/master"},"bd9f09ce8fa9c246f68ec8a89babc5ffb4657e09":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-16 12:24:25.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/5"}}},"commit":{"parents":[{"commit":"21fbc91df92af0310c5c829974b1b7b95259a4c6","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21fbc91df92af0310c5c829974b1b7b95259a4c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 12:23:56.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nAlso, This patch remove unused param context when\ncreating device profile.\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bd9f09ce8fa9c246f68ec8a89babc5ffb4657e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bd9f09ce8fa9c246f68ec8a89babc5ffb4657e09"}]},"branch":"refs/heads/master"},"c4bc275383448052e9aef9c87927498c4cefdfbd":{"kind":"REWORK","_number":6,"created":"2020-01-17 07:37:25.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/6"}}},"commit":{"parents":[{"commit":"cf548870f06f07a367309ef93e61eba95032c578","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf548870f06f07a367309ef93e61eba95032c578"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-17 07:36:53.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c4bc275383448052e9aef9c87927498c4cefdfbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c4bc275383448052e9aef9c87927498c4cefdfbd"}]},"branch":"refs/heads/master"},"5ef08166bdb20bf8992d047398b3a6557c662501":{"kind":"REWORK","_number":7,"created":"2020-01-17 08:51:11.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/7"}}},"commit":{"parents":[{"commit":"cf548870f06f07a367309ef93e61eba95032c578","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf548870f06f07a367309ef93e61eba95032c578"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-17 08:50:37.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5ef08166bdb20bf8992d047398b3a6557c662501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5ef08166bdb20bf8992d047398b3a6557c662501"}]},"branch":"refs/heads/master"},"d461b01a654f6ae211c9fdcc1e69a404293aaa48":{"kind":"REWORK","_number":8,"created":"2020-01-17 08:54:47.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/8"}}},"commit":{"parents":[{"commit":"cf548870f06f07a367309ef93e61eba95032c578","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf548870f06f07a367309ef93e61eba95032c578"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-17 08:54:17.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d461b01a654f6ae211c9fdcc1e69a404293aaa48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d461b01a654f6ae211c9fdcc1e69a404293aaa48"}]},"branch":"refs/heads/master"},"8d88f22757b5c6da00a8fe4dba1860f2593f0ca2":{"kind":"REWORK","_number":9,"created":"2020-01-19 01:35:56.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/9"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 01:35:28.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8d88f22757b5c6da00a8fe4dba1860f2593f0ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8d88f22757b5c6da00a8fe4dba1860f2593f0ca2"}]},"branch":"refs/heads/master"},"6b28ca2722885f4eb4e0ae40b26b19d2347e21c0":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-01-19 07:32:18.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/78/702578/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/78/702578/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/78/702578/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/78/702578/10"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-14 03:30:32.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 07:31:20.000000000","tz":480},"subject":"Improve UT for cyborg/db device profile","message":"Improve UT for cyborg/db device profile\n\nThis patch add some UT for device profile:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: Ia6f415b74150cdda94d71dc8b5e760f7758a526e\nStory: 2007091\nTask: 38120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6b28ca2722885f4eb4e0ae40b26b19d2347e21c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6b28ca2722885f4eb4e0ae40b26b19d2347e21c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
