)]}'
{"id":"openstack%2Fcyborg~702825","triplet_id":"openstack%2Fcyborg~master~I5e683c99d1e08ed6a166a110a87b665cdbc5bde3","project":"openstack/cyborg","branch":"master","topic":"improve_ah_db_ut","hashtags":[],"change_id":"I5e683c99d1e08ed6a166a110a87b665cdbc5bde3","subject":"Improve UT for cyborg/db attach handle","status":"MERGED","created":"2020-01-16 09:06:29.000000000","updated":"2020-02-12 13:16:55.000000000","submitted":"2020-02-12 13:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702825-1581513330621-7c055bf3","meta_rev_id":"3c405ceba425e5ed9c197983731026a3c5b4611b","_number":702825,"virtual_id_number":702825,"owner":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 13:15:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":2,"date":"2020-02-12 09:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2020-02-12 12:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2020-02-12 09:37:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-02-12 09:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2020-02-12 12:34:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 07:51:35.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-01-17 07:51:57.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2020-01-17 07:52:00.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2020-01-17 07:52:05.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-02-07 08:06:54.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-02-10 13:02:35.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-10 16:32:07.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-02-12 12:34:42.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-02-12 13:15:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35215b7612cc6d093b77492b90bc12084656189c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-16 09:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4709d949dea00b1b01f4ab450b6e28ca535c3205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 09:53:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9ad4dbf981046419be6fa3960a51b0b : FAILURE in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf5bc5a25d24d51946d11de8157037e : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2229e6d0b29f4fe88231ed2a103e52fc : FAILURE in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d39e0152e564c4d993122dc434fb111 : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4ffa4314544ad281ff71c071eae202 : SUCCESS in 3m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/06f0030bddc64d3aa5d1d58f203b51d7 : SUCCESS in 31m 24s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1271a45c5ff0462ca37f32bb93b2ec22 : SUCCESS in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58b68df724fc5bde74e141b751697cdd9ca0603a","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 03:57:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0832eaf972a2b54c3bef18bf1e14342c67eb88f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 04:39:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d43ba03a682a44e2b809f7e73ec09e84 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ec532a52eb424cb4f1cbb3b2c2f37e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/398835a55fce44dbbb0dd8925c2dee80 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf002902f77149fa875c315202f81f54 : TIMED_OUT in 41m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c952f6941a6947a8b069e3fbc7d49358 : SUCCESS in 4m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/16a41d2130d94ad9ae6212cc804ba8b8 : SUCCESS in 32m 38s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b5630be495b4df8b8f0fc874be72415 : SUCCESS in 32m 51s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/f7cd216fb4004e68ac84fdc4b7969d39 : FAILURE in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5df846ca94eed06cdf148d885dfba22f5f951816","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 07:49:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"86a53c4df9be7cc7a95f54a7b308d690a048172d","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-17 08:07:54.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"126efb8ef927f935eb972ae312c4e34d8208fae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 08:51:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bcc61697c2b48dfb3b8da0f00884485 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952fd220789d4e6c8aced192353f1666 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea79f14205d411aa1e84d6c291bc126 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffb2b1bf5701409d8d44c57da9918d9c : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5eca394f1c943e7917dcc9f1ec140e8 : SUCCESS in 5m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1365d0aa6d0d401ba0a61935710e6840 : SUCCESS in 35m 15s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/568571d6a0de454a9c14747051a845e2 : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e886699200c3fed8a7a45b2897424811ea8b1ed","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-18 17:35:32.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd37a8fd8d8a84620e5c3a677b0478aa15566c3a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-19 01:42:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nsoft -1, a nit inline.","accounts_in_message":[],"_revision_number":4},{"id":"d39d4b7668746fcbebc8eb7a60f988798037db7e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 01:54:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fa051fdf84127096b4676b7a877365fe848f19b5","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 01:56:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d54bbda6f1f05dd8ae36ea076129fd8efb76033","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 02:07:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bb833c2e58bae549a57687bf6974405ea07786af","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 02:47:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"678ed65299aab30f871cb174d086cd18231a0744","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 02:50:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b7ee06480a76f9cc822e1df299237d8f9b2e3225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 03:27:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d9d048f04734270a153cc48acfaaf95 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01789a33ca0244e082306cebdd691f92 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbe3da47549e4b458af1c8d71c307e88 : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3945e08da6364a8d8b59aa01f966b0bd : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bc0acf883a403aa11d7d3548743e7c : SUCCESS in 3m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f63b941471464c3e844df2b97d88490b : SUCCESS in 29m 40s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b2f27778ec04413b7eb6cbcd79aa031 : SUCCESS in 31m 50s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/2413f59d9bba47128018e1e8d91792f9 : FAILURE in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d59928dc00a724d40b4224e0b290757e19a5cd32","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-21 09:12:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"761f9a3f155f82e4dfed06421688750cc3feadf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 09:59:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beba060c1b0b4766840b9c12a25e6f1d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a82d9f5e8a4dc1b9f59a451e6ad996 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e30a0a3907e04ac7bdb193eed0d2dc50 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfaf131cc790429292f267a48beb82f5 : TIMED_OUT in 41m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/989f683f7d704136bec5be51e6f1260f : SUCCESS in 3m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/147ca4c952b8469498a3f79f76cfbc99 : SUCCESS in 32m 03s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4db62c3c757347c297f98f1895d7d3d8 : SUCCESS in 33m 35s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/aa1838a4b1124e80b75e9f46b2217b56 : FAILURE in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"86409f5cb89bca98645cf3d40fdd318afcb67f92","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-03 01:20:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d2148aeb10c438c570f3e2d732117c475cdfc12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 01:49:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10daa38d1cee41ebb688f00817746e36 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2345a5fab5b541e8afe5cdbc29cae89e : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9cfbd2fb39f44f7b68587668912f6d8 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/261c3c15e216449c971879c81cce6e89 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1338c29eca264380838d01d0369e46c2 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea72b28e56f43b6b671fe23ff7bed5c : SUCCESS in 3m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f9acfd5ad2574fe8a986e80c1b8e8b24 : SUCCESS in 25m 45s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93d8ca77353446b4bf77ba4bbe0194b9 : SUCCESS in 27m 49s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/85b2b877393b4724a2720b15415c3eec : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"59147c876399069c965e937648a212e5e5cf4997","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-02-07 08:06:54.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"434444164d50364be52a4a3a38c6b6ec92b77b55","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-08 06:41:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8e262caedf169c1c4a7399482472b43e408a25c1","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-08 07:27:01.000000000","message":"Patch Set 7:\n\nWill do it. After the parent patch be merged.","accounts_in_message":[],"_revision_number":7},{"id":"9e76fff5797d6037fed27ecc68e22d218a669638","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-09 14:12:22.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"d24e30fabfb927403b46c203ad5a4f211a58ef94","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 02:30:20.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"af4aa84454c068c4b4831b5a659beff1307a86ad","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 02:31:29.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b46c8a279e7a183cf9f725db4a73d1d391595593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 03:16:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b334ad3564e40b280a2bd5f2d4eb1aa : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0107e0348edc4351852f56d899f77898 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fe3c04f8be14348a56322225a438846 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc5d45cca8d348b2b1acfc77931701df : TIMED_OUT in 41m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45caec96de0f43f79854ab087ff9af94 : TIMED_OUT in 41m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f72babc021f43f4b5404f828de0a161 : SUCCESS in 3m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d5d4f3a2f886490e83f514ab4bd127be : SUCCESS in 33m 48s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ada2b4b7fdd45cd9065229d4000d03d : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"63bb9204a5787abd0bbca7fd7c322004d41797f1","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 03:22:47.000000000","message":"Patch Set 8:\n\nrecheck py37 timeout","accounts_in_message":[],"_revision_number":8},{"id":"63a899f0e1eb27c76f1e1df8c11920cdfbfe60d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 04:04:49.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/112221c437b74ed6823a9de6ffdd064c : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe4f12d2afe43689d2b6251d7ad4006 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a1b709139cd48d3b859a06510468856 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8cba7e788044be683408af1dde674bc : TIMED_OUT in 41m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/229819a4a08843f2b946b9c90ae7ad1e : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33692200f464967b81b9973a55e5874 : SUCCESS in 4m 06s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c6a7b854fef14013b561465b8cfc6e7e : SUCCESS in 30m 10s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0346f3d12044a18a425e7239ffc9927 : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e37d5d271723f4fe35dccd7c12bd93f02a624806","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 05:10:56.000000000","message":"Patch Set 8:\n\nrecheck py37 timeout","accounts_in_message":[],"_revision_number":8},{"id":"bf98ace2214bbd6dbd3d52fca6493729c1be93ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 05:45:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82233bd130f0414bbad2ab0625b3438c : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1605f8128045478e89ac6ca639ffd509 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/309ec9ccd7ec4395bd63b2c24a7c0c63 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3284105da09e48da84157ecbaf57b597 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c63816ca414f2e9cd3b9f19dce2dfe : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ea23f5fb854c34ae9d5acc82921b47 : SUCCESS in 4m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/83b039e927fc471aab6016fc1392a6e2 : SUCCESS in 27m 39s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1442d25a596c48bb8f79ee02f5487027 : SUCCESS in 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4530a2409f3e363ddd10574507e4c9767cfb8247","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-10 13:02:35.000000000","message":"Patch Set 8: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"e3b870e6a684a4bfb6848a06a0b50181d0eaa079","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-02-10 16:32:07.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"112ee6951209dce0dcb7b57561a551140eaead06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:32:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"525a4c62aa7407b9e0f65c2a0fc952b5349275ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:14:55.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1db12b6d2d407281b1b0da33a637bd : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/091eae6cce5744fb98563b27c8cda600 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2895d12e665412db8fb9cbc60c00667 : TIMED_OUT in 42m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d28089c5d1a47c4aa8a840a7fddc337 : SUCCESS in 5m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/40ce0d60f5fd4496970f6519f527907d : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":8},{"id":"515d9fd887832ae614e90aca468e82d81bdd7f8d","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-11 00:54:49.000000000","message":"Patch Set 8:\n\nrecheck py37 timeout","accounts_in_message":[],"_revision_number":8},{"id":"b271bfb323715f7687e405430c60a24398765ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 01:39:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63e5fb58cae04884b44c3fb2ca6f7fd2 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391c6c299cf34a8e81fc03bab874ec5c : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2334a05cf64ea195a3bed78e8cdbff : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea06daeaa1b047e29741b0aca0de31da : TIMED_OUT in 41m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf52fc65a7d84f34b8a3fd6c1333adcf : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bba7f1cbf4e4982aeb6bc2180cfd1cd : SUCCESS in 5m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ebdad889d6964596b80ed9f7005de130 : SUCCESS in 37m 27s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ec4e4b4e4dd43d4a1517107b884be22 : SUCCESS in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"72072c30aafd55f83271d07fe7fcec342e00e7f9","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-11 02:06:38.000000000","message":"Patch Set 8:\n\nregate py37 timeout","accounts_in_message":[],"_revision_number":8},{"id":"02a599899b1e07111ceee7824c2d7ab3f73c09fe","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-11 02:47:49.000000000","message":"Patch Set 8:\n\nrecheck py37 timeout","accounts_in_message":[],"_revision_number":8},{"id":"61cb1fb32dc511c6e5ba6b49a9c5f87fe82cc4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 03:32:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e094af8987cf476884f60b74ceeae5a6 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc603e0b8c6481c8cfdd4d9b11016f8 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/343aa8de702c4be8a9061675fc52545f : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba869a1321eb4424ae9d96ef94d245e3 : TIMED_OUT in 41m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c5bb2aefff24644adc1fd2620d407c6 : TIMED_OUT in 41m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce41d07295b4276be1fe836e397acc5 : SUCCESS in 4m 59s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5cea89522f6941b0bcba50f3519583dc : SUCCESS in 32m 14s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9673497a9f94a6e969b9f8c144bbbb3 : SUCCESS in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"edd10b6b1665ea19525734d4f487add585b9678b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-11 04:19:40.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"59896f2e0b6cccd950634702fe00ac57faff3521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 04:55:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67911fad362a4b7e9fca23731b4f2b1e : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54f2d3d18cd44df8c417fa8d678b2e9 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f98efd8f845647ac82ba43c61cf0c685 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cc294d905924368a8a7795932cfd499 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7842f56e1f11411fb26e4ddc6aada4bf : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1a423d754642b7947dac802bb8e752 : SUCCESS in 3m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0cd497261749490ca116138f45a177ce : SUCCESS in 31m 22s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00ea1109a76a4ce09243ce5968de9e70 : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"208e40f2b6114b92880ce6466caa56f7b246f1b5","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-12 01:14:37.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"38b760252a138dc85c8d739fa2a0c0d807bb8356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 01:52:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb72b5ada775489cba70b011872d0ff3 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4e3943439e4e9293c0c17cede833a4 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b69c1cfc53042e0812b214cf7ca495b : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0e789bfe2794fd3a688d184d582e304 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70217197dcb14d8a8a7134b2f15218b7 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da7830d319e44b5891281d97c6347d5 : SUCCESS in 6m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a7b35e76cd6543e9900f2ad13ab55871 : SUCCESS in 26m 40s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3abec6987a24e9b8607b5aabe348395 : SUCCESS in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d583028c4943c8ec2de7443eebf9460831008796","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-12 02:23:42.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"c3bdc46e68dbdf86d5c2ceaeb83c809fcf969d60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 03:09:11.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6549413b3bd948ae9ae3f957655045a2 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872517d6f51442719af01b78df12f474 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/840e6f624d4d4ef28fa177c9b7967a61 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29ebde3882c34c6a9b4c43e901730474 : TIMED_OUT in 40m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b92637f57ed418fadb88c17208f0575 : TIMED_OUT in 41m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a407daca830541bfa1f5863368229509 : SUCCESS in 5m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ea43bd56e8a4411880e5a1cf3e93aa81 : SUCCESS in 33m 49s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4414a16c11404dba90778521f3b11fe3 : POST_FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"09b593857e87227eed1e379412cd0a4923854fab","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-12 09:37:40.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"156282b38340f0561f5f0f20b3270c6a7d8b94f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 10:15:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48aa7c601f2b4606a366b1617693671b : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f3a62fb38b34c14ab3bfc320defb095 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7caec0184a4b488089df52ae2b0766bc : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/601cdbc7531442b3af9519f77ccedc18 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3fd51ce9638410a929d432efcff9484 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae153399eccb4bb282d81083d2e1407f : SUCCESS in 5m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/15c8024fefff417a8506ad0eaf833478 : SUCCESS in 31m 45s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbacc29f1e7347e593bb2fdbf08f83e4 : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d0da40080da2a931a3bfcabe20adf69b79c542e6","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-02-12 12:34:42.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"972025a350e847e2d7fe9487fb70efc80284cd98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:34:54.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"9db731dc5a5b2016f9d9d5ac33d46bf15672b19e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 13:15:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"fa8927860ec9f4e0fba9aba0d865982dd1ae235f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 13:15:30.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4538a423db194a0da9a18c8cddb1b778 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac08723689c4fa2a08434e199481edc : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38935880326e47c9a465131945f1f3ca : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d222871c7640778b1fa90a8f8171fe : SUCCESS in 6m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5ce3c2af7f7b455086b9b9907f90b825 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":12},{"id":"63af09a3a3718d516484fbd345edcedb02871d08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 13:16:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c17b57e49d48469001bb6017d9de95 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"e2e1e3f156d556becfa9cd5edf2bdf0ebb8f9214","revisions":{"04d3b1feba3a959b922d2c6adc3195fcc6192801":{"kind":"REWORK","_number":1,"created":"2020-01-16 09:06:29.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/1"}}},"commit":{"parents":[{"commit":"329e5f3b733e42f88ccad80c8624ee72a63a5814","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/329e5f3b733e42f88ccad80c8624ee72a63a5814"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/04d3b1feba3a959b922d2c6adc3195fcc6192801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/04d3b1feba3a959b922d2c6adc3195fcc6192801"}]},"branch":"refs/heads/master"},"f1e4f0a942c53d49752581f39191dab82870a30f":{"kind":"REWORK","_number":2,"created":"2020-01-17 03:57:38.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/2"}}},"commit":{"parents":[{"commit":"21fbc91df92af0310c5c829974b1b7b95259a4c6","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21fbc91df92af0310c5c829974b1b7b95259a4c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-17 03:57:03.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f1e4f0a942c53d49752581f39191dab82870a30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f1e4f0a942c53d49752581f39191dab82870a30f"}]},"branch":"refs/heads/master"},"dfb2e0254ed4bf367a3c74b157d6d5894c756a1b":{"kind":"REWORK","_number":3,"created":"2020-01-17 07:49:22.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/3"}}},"commit":{"parents":[{"commit":"cf548870f06f07a367309ef93e61eba95032c578","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf548870f06f07a367309ef93e61eba95032c578"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-17 07:47:31.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dfb2e0254ed4bf367a3c74b157d6d5894c756a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dfb2e0254ed4bf367a3c74b157d6d5894c756a1b"}]},"branch":"refs/heads/master"},"b33711d40194901e4c556598491402a88ecaef12":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-17 08:07:54.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/4"}}},"commit":{"parents":[{"commit":"cf548870f06f07a367309ef93e61eba95032c578","subject":"Public interface define [Improve UT for cyborg/db]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf548870f06f07a367309ef93e61eba95032c578"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-17 08:06:06.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b33711d40194901e4c556598491402a88ecaef12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b33711d40194901e4c556598491402a88ecaef12"}]},"branch":"refs/heads/master"},"818ea69ee9aa1ceac5d9a773a95853f1e7dc5580":{"kind":"REWORK","_number":5,"created":"2020-01-19 02:47:59.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/5"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 02:47:10.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/818ea69ee9aa1ceac5d9a773a95853f1e7dc5580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/818ea69ee9aa1ceac5d9a773a95853f1e7dc5580"}]},"branch":"refs/heads/master"},"84c59f160095d430faccd10cb9a913afee407999":{"kind":"REWORK","_number":6,"created":"2020-01-19 02:50:43.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/6"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 02:50:09.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/84c59f160095d430faccd10cb9a913afee407999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/84c59f160095d430faccd10cb9a913afee407999"}]},"branch":"refs/heads/master"},"ab329d89c3f8a42ad06904ea71b49b4b4bd91474":{"kind":"REWORK","_number":7,"created":"2020-01-21 09:12:07.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/7"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-21 09:11:38.000000000","tz":480},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ab329d89c3f8a42ad06904ea71b49b4b4bd91474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ab329d89c3f8a42ad06904ea71b49b4b4bd91474"}]},"branch":"refs/heads/master"},"165d5fbe56593ee6d0fb246a4574cf24fe64f3e6":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-02-10 02:30:20.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/8"}}},"commit":{"parents":[{"commit":"acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a","subject":"Enhance the db layer to verify filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-10 02:30:20.000000000","tz":0},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/165d5fbe56593ee6d0fb246a4574cf24fe64f3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/165d5fbe56593ee6d0fb246a4574cf24fe64f3e6"}]},"branch":"refs/heads/master"},"9a3c317a5af06eddc905165990909f144b7a6627":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-02-11 04:19:40.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/9"}}},"commit":{"parents":[{"commit":"d6e39af21d88191f17ca823d006ee116aa46ec45","subject":"Solve py37 timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d6e39af21d88191f17ca823d006ee116aa46ec45"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-11 04:19:40.000000000","tz":0},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9a3c317a5af06eddc905165990909f144b7a6627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9a3c317a5af06eddc905165990909f144b7a6627"}]},"branch":"refs/heads/master"},"fc8637de264672d5e73e6464450019039fddebfb":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-02-12 01:14:37.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/10"}}},"commit":{"parents":[{"commit":"5b6f26abb8b474276adf6a298c5b8ad0aafa413f","subject":"Guess for the root cause of timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b6f26abb8b474276adf6a298c5b8ad0aafa413f"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-12 01:14:37.000000000","tz":0},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fc8637de264672d5e73e6464450019039fddebfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fc8637de264672d5e73e6464450019039fddebfb"}]},"branch":"refs/heads/master"},"50b39878dc7e3bf14848386edba73c0804e4d0f0":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-02-12 02:23:42.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/11"}}},"commit":{"parents":[{"commit":"acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a","subject":"Enhance the db layer to verify filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-12 02:23:42.000000000","tz":0},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/50b39878dc7e3bf14848386edba73c0804e4d0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/50b39878dc7e3bf14848386edba73c0804e4d0f0"}]},"branch":"refs/heads/master"},"e2e1e3f156d556becfa9cd5edf2bdf0ebb8f9214":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-02-12 09:37:40.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/25/702825/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/25/702825/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/25/702825/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/25/702825/12"}}},"commit":{"parents":[{"commit":"5b6f26abb8b474276adf6a298c5b8ad0aafa413f","subject":"Guess for the root cause of timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b6f26abb8b474276adf6a298c5b8ad0aafa413f"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-16 08:39:12.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-12 09:37:40.000000000","tz":0},"subject":"Improve UT for cyborg/db attach handle","message":"Improve UT for cyborg/db attach handle\n\nThis patch add some UT for attach handle:\n1. get\n2. update\n3. create\n4. list\n5. delete\n6. allocate\n\nChange-Id: I5e683c99d1e08ed6a166a110a87b665cdbc5bde3\nStory: 2007091\nTask: 38161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e2e1e3f156d556becfa9cd5edf2bdf0ebb8f9214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e2e1e3f156d556becfa9cd5edf2bdf0ebb8f9214"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
