)]}'
{"id":"openstack%2Fcyborg~703281","triplet_id":"openstack%2Fcyborg~master~I39b3f02e898b67e4d4eb686b5a6cf9065c6280de","project":"openstack/cyborg","branch":"master","topic":"improve_deployable_db_ut","hashtags":[],"change_id":"I39b3f02e898b67e4d4eb686b5a6cf9065c6280de","subject":"Improve UT for cyborg/db deployable","status":"MERGED","created":"2020-01-19 09:11:23.000000000","updated":"2020-02-13 07:30:24.000000000","submitted":"2020-02-13 07:29:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703281-1581578940459-0cc33de8","meta_rev_id":"8545b92948690de9860d6d754e9574de473beada","_number":703281,"virtual_id_number":703281,"owner":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 07:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-13 06:15:29.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":2,"date":"2020-02-13 03:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":2,"date":"2020-02-12 11:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":1,"date":"2020-02-13 03:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-19 09:12:15.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-01-19 09:12:34.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-01-19 09:12:39.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2020-01-19 09:12:42.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2020-02-07 08:12:12.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-02-08 05:28:12.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-12 11:46:05.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-02-13 03:49:02.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-02-13 07:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"916659537be38becbf4782ddc2ee335b997b7203","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 09:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08bed1ec19bc03341089c3b38bcbdb5171b8ed26","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 09:14:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5952673f0a69e4f343b12895df1f7ec6ca9a0143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 09:49:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/061d5274ed274ece8bcbfadb39cfe7f8 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44fcb0c5d8214d90beeff1ed552ca829 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3c22fd91e4e4784bf783e68953ba039 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33358e775e2f4a5cbb87b23ee849d4f1 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538dc83f680d4cba9027eae4bcd2a589 : SUCCESS in 4m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0e7bd613e19a4cc39783ea620a1518b1 : SUCCESS in 29m 51s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06f5b21063a74a3a83383de049849969 : SUCCESS in 30m 34s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/991c28d6bdb04489ba5765d8a3b01cbc : FAILURE in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e121f79d0bef2e525370454936e046da4cbd8df5","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-20 07:27:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a892f06b340999b7c916e57f65e83fe72df88119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 08:02:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58f9bc376d2e42d3b08cc7954482eae9 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587712f0f793462f8a2ad6d8515dc920 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace1d92826324980b2373ed2f0b4395e : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1899c98cd784431bf47b690ebc3a5ec : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896dc10f974249a7ba90175c5c1ec47b : SUCCESS in 6m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0f560442840c481487261194eb2dba49 : SUCCESS in 27m 43s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d576a180a07410fa2847802d80ba80c : SUCCESS in 33m 03s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/8c002973b7f349bd8a0c66259eddb7e5 : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a2ad0d7cc6a2168c169324ea5512e91fabc8a3e0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-20 09:21:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a029cfa97000f44f3e36351a67c57a6045eedb85","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-20 09:43:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nThanks chenke, please see my comments inline.","accounts_in_message":[],"_revision_number":3},{"id":"b5bba7379a69946a8ef3fe06fb35fc108616db6c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-20 10:01:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ef046cd158650ad0f697c144f0a3766e2966e71","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-20 13:46:03.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6e07f550a8255e1e2279745a7cf3c32c5b8b7567","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-21 02:44:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e4741ee1425ac3a927bc8141b1604e27990b893","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-21 02:56:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac75b358d0d0510fffc1239fa110ef9232f9f33d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-21 06:16:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c847717cea882d39c40af13d537857f85964e311","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-21 08:52:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"178c9b923e593618cc3826ce0a26cabfbe8e8df8","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-21 08:58:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b24fc13767f3e2bcc17df5414868f8655f1a2129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 09:24:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97efd193073242d6bdb2dab28d9a7b12 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150d44172df844a7ae3d89539dee5767 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4831876ad25443ab8fa203e86d64f51e : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76926d4153a744fa8f09a12f8d922c5a : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10db673dfd3d4591ada886ac4405f7ab : SUCCESS in 5m 54s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9bae1813b9104f40aa8193524a7e1374 : SUCCESS in 30m 31s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aea4fed1f3b4993b2b2a62ad2aa3cd0 : SUCCESS in 30m 30s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/dfcaeb626e744790aa123a42a704938b : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"faba262f1e37595ef567044ab6bce9bde13c0dea","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-02-07 08:12:12.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, please remember to refactor the fake data file later.","accounts_in_message":[],"_revision_number":4},{"id":"f88f0d20f7a30981df11e3f384a136b3393ec2eb","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-07 08:17:41.000000000","message":"Patch Set 4:\n\n\u003e LGTM, please remember to refactor the fake data file later.\n\nThanks Xinran. Sure. Will do it.","accounts_in_message":[],"_revision_number":4},{"id":"68d757afb58e1b27e94da7c2d0f819f5f274d406","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-08 05:28:12.000000000","message":"Patch Set 4: Code-Review-1\n\n\u003e https://review.opendev.org/#/c/703049/ already fixed this(because \n \u003e it depends-on https://review.opendev.org/#/c/702578, so it was not \n \u003e merged), I think you should rebased on that patch, and then run \n \u003e your test.\n \u003e\n \u003e You can:\n \u003e git review -d 703281\n \u003e\n \u003e and then:\n \u003e git review -x 703049\n \u003e git commit --amend\n \u003e\n \u003e finally:\n \u003e git review\n\nI saw you are not reabse instead of combine that to this patch.","accounts_in_message":[],"_revision_number":4},{"id":"8a1824a7835dbc2997e860cdcf807265412b033a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-08 05:45:33.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nPlease add a releasenote for removing the db api deployable_get_by_filters_with_attributes().","accounts_in_message":[],"_revision_number":4},{"id":"6599ad03a3a1901ad07e3316223e2d7b64f72004","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-08 06:20:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d4817f1e9ad702991f9d8333e485e267b2c31eb4","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-08 06:39:26.000000000","message":"Patch Set 4:\n\n\u003e I saw you are not reabse instead of combine that to this patch.\n\nAfter the parent patch be merged. I will rebase it.","accounts_in_message":[],"_revision_number":4},{"id":"bf16d0be2adb6f8dadc0f3c20d6b13c98ed6c490","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-09 14:12:46.000000000","message":"Patch Set 4: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ac0860205355d0220651e4c852497cb0d24d95f1","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 02:59:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cfb5769f3961055835a3352ea7d2caf864f09da2","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 03:06:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc51adeb5bd064843f4d3196bd90eb3f7417083a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 03:41:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/299aac57167c4dcfae85def7d8e6dbea : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8f7696d1b341618078a4c2702262d1 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eab9dc1f768944c1a9ee49eb3f6c33d7 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12508ea5820646929e48954c54a09271 : TIMED_OUT in 41m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/788ec507940b4fcbb985c0a99f4a9830 : TIMED_OUT in 40m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1173756cbacd474c971a17da119f0741 : SUCCESS in 4m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/92a24f9b35e04ed883789fe66cb67fb5 : SUCCESS in 27m 58s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63625e75dc00464d9072221230c887a4 : SUCCESS in 29m 21s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/f31dd12793b8441182a50eb2e3037e0f : FAILURE in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"64929db0e452119d586b80027e9a6e00df4c1772","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-10 03:44:05.000000000","message":"Patch Set 5:\n\nrecheck py37 timeout","accounts_in_message":[],"_revision_number":5},{"id":"d53efe50efe1826e2f17146c77e6cc18b8fd914b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 04:29:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9aee34db6eb492aa59d323f41bee10c : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4664dfa705074a60a63ade77d63667be : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13fd115a02384a3a9519661f860de691 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b33ea37457db45cabfb4e2dec471cc00 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8089d08e67e641699dec0b8d603d5bc0 : TIMED_OUT in 40m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f83de81c974b53984d6358316d1a8a : SUCCESS in 4m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/64eb0ab8ea17483989e6ad8f0d44d6f9 : SUCCESS in 28m 49s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9961811ff072457a944b01abbd666f5e : SUCCESS in 30m 47s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/b1208e669b794aa59caef37745fc6054 : FAILURE in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4f7a32f2cffc6044b540867dff95736686824598","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-11 04:48:19.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d31f3025d1ca11a95a42b6804960446465ecc60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 05:19:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdde3743d1df4d5eb45c67960b501a9a : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f76c115e114e5e9dffdfe48c8cfbfb : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/166205abbc1843ecb3d4e5af2b3bac34 : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6774d4b012904de8a8d26f7df51dcaf1 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03551b21b9fa4208888b0160bcecad61 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38ec3ab92894c7e93d1ce9585f2756d : SUCCESS in 4m 33s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d64f3b170aae4b488286cdeae6877731 : SUCCESS in 27m 17s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c5f6a35bba44dafb849ced0daf222b2 : SUCCESS in 29m 07s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/441d5deef4604e358dcfb768b2a3ea89 : FAILURE in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"74ae20a89c95949b34afe75c248637f7e3640c71","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-11 15:01:20.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f2d49926a240475a8f2d6f201f45f335ea2d659c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:36:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f27ea0046c734bea89144920bee1eb21 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3d45eb3c124fb6bce0135548717594 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52b43e67962a4fcdb750fcb6163bc1e1 : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/586eb0f24ecf41679fedb17a0979368e : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b08f2dee95f7406688989fa32276056a : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023ac591d0204c7581b85678c2906abb : SUCCESS in 6m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9225aca6a62e4fc1ab06a59e14b37f5a : SUCCESS in 29m 18s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37d0aa29bec448b5925a720d308e60e0 : SUCCESS in 32m 23s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/8160ae1a7e7a4ccea586b84fea66761f : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"92bbcdf28ad21d445be431e417c873f7d99802df","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-12 01:09:50.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"28d35fea2de47be185ffacb92bd84419e651f293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 01:47:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c38bce73dce0437384a09cd0e7d8ede9 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae8c41a032d486cabebc8a9db1ac4bb : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccb2b8412d4e41bcbd32e63cefc98319 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8bb6dcfb54a4e03a269851ecf298ab3 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6be17a970164ddc8ed8de4bec290849 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb20684ebbe84cc1bab13ddab5b45c3c : SUCCESS in 5m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7ad1e3930e9840fb8a8b86fe9afddad1 : SUCCESS in 37m 09s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c9f1fb0e29044ba9dde0d1fa32b1ce7 : SUCCESS in 31m 16s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/2b0269608c4f4d709b40371cefc14174 : FAILURE in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"736b586bf580041a6ac90fadec817ae31d937f06","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-12 02:24:37.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"423745ea2ca346df08c7156f22a0a8390b5650a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 03:10:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6130f206f7ab47adb0374f9cdd1be264 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43842f1b58a4f52b663354232b380d3 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62c921d383e54d7295edbb95cf3c40ce : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32870a40def8436c82758ce2c54bf310 : TIMED_OUT in 41m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191df2e6cab849179fc31a7e699f5502 : TIMED_OUT in 41m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f361434419474ca9b417b0e3ec37f9b2 : SUCCESS in 4m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e512d106bec14030a7c44eea80581e49 : SUCCESS in 31m 03s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73bf31d5b2604b7fab6bb6ac2e4487c7 : SUCCESS in 31m 49s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/62cec3b511fb4b02adb3216ba53a8e23 : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4cfe0eb85e527e734aa9fb3acbb0fdbe08705f27","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-12 09:38:06.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"36df16b1388030c3778e741254bc18baec13babd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 10:14:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff72538f7aa749c0bf8b1d3de0d84004 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbda05e69ffc47a5963917ec176c9658 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c0e26285c8347d696f116e2e1afec20 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62e79e2dab2243d195dd457822734dc2 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/311ea21b622c4ae9a713ec808cbc2994 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5563a52c9a84b68a4320d0d5ee4184f : SUCCESS in 5m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/825c185f2d0846e6ae1647aaaf2ac5fd : SUCCESS in 31m 59s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8948700331c24ed08c95ed0862b6b6f9 : SUCCESS in 33m 41s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/3601d756b3e14636b28854d3c0ca9d5c : FAILURE in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dfdab9de1d27ee7975216374c80b06e80fcaf97a","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-02-12 11:43:45.000000000","message":"Patch Set 5:\n\nLGTM.","accounts_in_message":[],"_revision_number":5},{"id":"2db7ae62eebca92f9389563fcbf2d318139c5303","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-02-12 11:46:05.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks chenke,good patch.\nAnd pls don\u0027t forget to keep an eye on the time_out problem","accounts_in_message":[],"_revision_number":9},{"id":"f3ded25dc1547895ad49042b7e7f4a9b1376986c","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-02-13 03:49:02.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nThanks hard work on UT.","accounts_in_message":[],"_revision_number":9},{"id":"33a0d40050f1376e546c41b2d5a46c3a0779bb87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 03:49:14.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a085ee92175a17ea85b8dfe4175a6696586aec85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 03:56:41.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae49e8fd260b4cae851554e3492f9c60 : RETRY_LIMIT in 1s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/980ea97388934156870f3146d09d4229 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ecc01bc069747b7ad83f9e99d82a4fd : RETRY_LIMIT in 1s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ede10ed9d0d42d58dc56f5d6416249f : SUCCESS in 4m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/65a3f818537648fb8b1a00d0a472174e : RETRY_LIMIT in 13s","accounts_in_message":[],"_revision_number":9},{"id":"87af871bb24e910e5404e76366372ff33f25f91c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-13 06:15:29.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d3e6a9de9c0dabe8cd8878e3e64c20829a998e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:59:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7e0b648f0514287b981c4121c398422 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a223e8030f40368202f5648a173553 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6fdfe2c36f4776b8035ba419746e00 : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7216796d51c4a9c9c75ce3760758d03 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd007cabff4047f3b9d7327d851c38de : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c26e6fcabb942778125fa6b12020890 : SUCCESS in 7m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6b2698d81f3749e5b91f7677cc5d7426 : SUCCESS in 33m 05s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a7b7773e8d04c7590ecbad814f65488 : SUCCESS in 37m 46s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/523ce769500340c98361d4be1f302c73 : FAILURE in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6f20a41ab5d84e401a24f4af88db492fab594438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:59:25.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d00727b041bde65114263b80a9b823906283fc6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 07:29:00.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22061a73ab544162a41fa7593859004a : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fab20ca8de442a1ac289f15ad3ae0d1 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/738543783d6e47ee8d6a427b1495d240 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a4a5dcf7914f04a0a03d5be14378c9 : SUCCESS in 6m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2c81a43b29f14b458f3a1a9b404c726c : SUCCESS in 28m 46s","accounts_in_message":[],"_revision_number":9},{"id":"617efdc20f938365f06ea1f9b981c3c743233fb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 07:29:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"05199a782ca38815d914fe1b97f147047ba385a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 07:30:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea7f1d4fd2f40a8bdc8308c23e7e1e5 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e7c6783858f8a48ac77966d7081056d19423c8a6","revisions":{"a9db20a97b7a320a009ae7529f898e791f1dc754":{"kind":"REWORK","_number":1,"created":"2020-01-19 09:11:23.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/1"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:10:17.000000000","tz":480},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a9db20a97b7a320a009ae7529f898e791f1dc754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a9db20a97b7a320a009ae7529f898e791f1dc754"}]},"branch":"refs/heads/master"},"516ff17b54ed10025da6262e9f9ac8a8e1a9ad09":{"kind":"REWORK","_number":2,"created":"2020-01-19 09:14:18.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/2"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:13:50.000000000","tz":480},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/516ff17b54ed10025da6262e9f9ac8a8e1a9ad09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/516ff17b54ed10025da6262e9f9ac8a8e1a9ad09"}]},"branch":"refs/heads/master"},"623140fd3b08be8ea0ff0d3a696dc6fa9e4a358f":{"kind":"REWORK","_number":3,"created":"2020-01-20 07:27:11.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/3"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-20 03:38:23.000000000","tz":480},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/623140fd3b08be8ea0ff0d3a696dc6fa9e4a358f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/623140fd3b08be8ea0ff0d3a696dc6fa9e4a358f"}]},"branch":"refs/heads/master"},"8829e17b4d29f55f5a086bcd9a99cfe7d7c02cf3":{"kind":"REWORK","_number":4,"created":"2020-01-21 08:52:34.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/4"}}},"commit":{"parents":[{"commit":"d8cbe092deaae6be2d90221ce64b40bffa3116c6","subject":"Define fake db objects for UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d8cbe092deaae6be2d90221ce64b40bffa3116c6"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-21 08:51:01.000000000","tz":480},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8829e17b4d29f55f5a086bcd9a99cfe7d7c02cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8829e17b4d29f55f5a086bcd9a99cfe7d7c02cf3"}]},"branch":"refs/heads/master"},"c64c6f608360298b3261e42d2de3e9b291c86ae0":{"kind":"REWORK","_number":5,"created":"2020-02-10 02:59:13.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/5"}}},"commit":{"parents":[{"commit":"acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a","subject":"Enhance the db layer to verify filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-02-10 02:57:34.000000000","tz":480},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c64c6f608360298b3261e42d2de3e9b291c86ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c64c6f608360298b3261e42d2de3e9b291c86ae0"}]},"branch":"refs/heads/master"},"33e40138c07b9f881e4e8ba32a6a98e8e147968d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-02-11 04:48:19.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/6"}}},"commit":{"parents":[{"commit":"d6e39af21d88191f17ca823d006ee116aa46ec45","subject":"Solve py37 timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d6e39af21d88191f17ca823d006ee116aa46ec45"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-11 04:48:19.000000000","tz":0},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/33e40138c07b9f881e4e8ba32a6a98e8e147968d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/33e40138c07b9f881e4e8ba32a6a98e8e147968d"}]},"branch":"refs/heads/master"},"435558fa2ff311269edefb9f5ae0fe59b5c75e55":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-02-12 01:09:50.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/7"}}},"commit":{"parents":[{"commit":"5b6f26abb8b474276adf6a298c5b8ad0aafa413f","subject":"Guess for the root cause of timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b6f26abb8b474276adf6a298c5b8ad0aafa413f"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-12 01:09:50.000000000","tz":0},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/435558fa2ff311269edefb9f5ae0fe59b5c75e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/435558fa2ff311269edefb9f5ae0fe59b5c75e55"}]},"branch":"refs/heads/master"},"d924ad1f1abbf12c6ba5816d675fe01e1aa277b3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-02-12 02:24:37.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/8"}}},"commit":{"parents":[{"commit":"acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a","subject":"Enhance the db layer to verify filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/acbc64f3bedc4f4946c5d12ea4fd7dc3d941667a"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-12 02:24:37.000000000","tz":0},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d924ad1f1abbf12c6ba5816d675fe01e1aa277b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d924ad1f1abbf12c6ba5816d675fe01e1aa277b3"}]},"branch":"refs/heads/master"},"e7c6783858f8a48ac77966d7081056d19423c8a6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-02-12 09:38:06.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/81/703281/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/703281/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/703281/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/703281/9"}}},"commit":{"parents":[{"commit":"5b6f26abb8b474276adf6a298c5b8ad0aafa413f","subject":"Guess for the root cause of timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5b6f26abb8b474276adf6a298c5b8ad0aafa413f"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2020-01-19 09:09:39.000000000","tz":480},"committer":{"name":"chenker","email":"chen.ke14@zte.com.cn","date":"2020-02-12 09:38:06.000000000","tz":0},"subject":"Improve UT for cyborg/db deployable","message":"Improve UT for cyborg/db deployable\n\nThis patch add some UT for deployable:\n1. get\n2. update\n3. create\n4. list\n5. delete\n\nChange-Id: I39b3f02e898b67e4d4eb686b5a6cf9065c6280de\nStory: 2007091\nTask: 38141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e7c6783858f8a48ac77966d7081056d19423c8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e7c6783858f8a48ac77966d7081056d19423c8a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
