)]}'
{"id":"openstack%2Fcyborg~708726","triplet_id":"openstack%2Fcyborg~master~Ied8ee91592eb0b4675f9c155e30a6c3a7df9b597","project":"openstack/cyborg","branch":"master","topic":"fix-report-failed-parent","hashtags":[],"change_id":"Ied8ee91592eb0b4675f9c155e30a6c3a7df9b597","subject":"Avoid creating a root provider when parent is not found","status":"MERGED","created":"2020-02-19 18:49:47.000000000","updated":"2020-02-24 15:52:30.000000000","submitted":"2020-02-24 15:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708726-1582559416889-dfe5b5ea","meta_rev_id":"d3b5bb4c8b013a74ac2bff08abc0e0f40e062c17","_number":708726,"virtual_id_number":708726,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"date":"2020-02-21 14:40:38.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 15:50:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2020-02-23 12:26:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":1,"date":"2020-02-20 15:53:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2020-02-24 06:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2020-02-20 15:55:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-02-21 09:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2020-02-22 10:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2020-02-20 15:54:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2020-02-24 06:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 15:53:43.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-02-20 15:54:56.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-02-20 15:55:37.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-21 09:05:33.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-02-22 10:52:12.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-02-23 12:26:54.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-02-24 06:42:59.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-02-24 15:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2331c4968dde5636241570256a40225f511f5c79","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-02-19 18:49:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a08f1d396582066a7f537725fcd638779e4e30a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 19:34:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e7abdec9c741fca05b419ef4888836 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2363252dac7e49c8a8b8b7d5a4c52158 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27c1de7b00742e5ad16fd3c912e2d9f : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25501d30293349f0aef64bd13b9c4aab : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/273b92589a9f4940b45336e653f31e27 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394478879dca4a15a368e58a5a8d6f58 : SUCCESS in 6m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/84cca021b1a24fd0bbcfa386f703e101 : SUCCESS in 33m 45s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ac464ab9f9546e983c5d23136bdbcbb : SUCCESS in 40m 32s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/6a059c010df54a3ea58fe4894285fe9b : FAILURE in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca565707e8e456d5cd3ab1650bd60773fb6db9fc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 15:53:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7b3aa7bbbddf8c295dfbaf5c8c451230c985852","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 15:53:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a2fb364298e2e647d63f011c738178f0d9bde59b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 15:54:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f72e6b63f2840e9e6b00e54a34b3516fa4537bb7","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-20 15:55:37.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"0f4f29408dbfa487760da545d0add8be6ec8a7cf","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-02-21 07:56:28.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a5f441cc0ea5dd792b2c94c652c7242f1a6e9d6e","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-02-21 08:17:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46974cdea39503fcb171e06789fedf557dd46ac8","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-02-21 09:05:33.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, the get_root_provider is before DB update/writing, if there is an exception raised during get_root_provider, the DB will not be updated/re-written. Thanks Dan.","accounts_in_message":[],"_revision_number":1},{"id":"e293b2b12de29d8bba7940ad3bb9d12ab5db5512","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-02-21 12:17:58.000000000","message":"Patch Set 1:\n\nXinran WANG:\nDo we agree Cyborg DB should not be independent?\nShould it never be updated if something wrong with placement update?","accounts_in_message":[],"_revision_number":1},{"id":"b8c670a12c1d15a64e25a49e0d31e8a25ed4b611","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-02-21 14:39:09.000000000","message":"Patch Set 1:\n\n\u003e Xinran WANG:\n \u003e Do we agree Cyborg DB should not be independent?\n \u003e Should it never be updated if something wrong with placement\n \u003e update?\n\nNo, Cyborg DB update should be independent with placement update. In current code, these 2 parts are in same functions, we should decouple them.","accounts_in_message":[],"_revision_number":1},{"id":"b7d06030273432aa8b14aefb0a14b35708d6b9fd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-02-21 14:40:38.000000000","message":"Patch Set 1:\n\n(3 comments)\n\n\u003e Xinran WANG:\n \u003e Do we agree Cyborg DB should not be independent?\n \u003e Should it never be updated if something wrong with placement\n \u003e update?\n\nI understand this concern, however, from our experience in Nova, it\u0027s much better to just consider placement as a core component in your project. Use it for accounting, don\u0027t duplicate its functionality locally, and don\u0027t let it get out of sync with your own data. When we were transitioning from our own resource accounting to using placement, we had a lot of issues where we tried to be graceful about placement not being setup yet or getting out of sync. In the end, we just had to write lots of fixup routines and tools for operators to clean up the mess.\n\nMy advice would be: if you\u0027re going to use placement for resource accounting coordination with other projects, do that, and don\u0027t let it get out of sync with you. (i.e. fail if it fails).","accounts_in_message":[],"_revision_number":1},{"id":"7c46f8241afd457c98d34b9724807f266ec26c2a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-21 16:07:26.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e (3 comments)\n \u003e \n \u003e \u003e Xinran WANG:\n \u003e \u003e Do we agree Cyborg DB should not be independent?\n \u003e \u003e Should it never be updated if something wrong with placement\n \u003e \u003e update?\n \u003e \n \u003e I understand this concern, however, from our experience in Nova,\n \u003e it\u0027s much better to just consider placement as a core component in\n \u003e your project. Use it for accounting, don\u0027t duplicate its\n \u003e functionality locally, and don\u0027t let it get out of sync with your\n \u003e own data. When we were transitioning from our own resource\n \u003e accounting to using placement, we had a lot of issues where we\n \u003e tried to be graceful about placement not being setup yet or getting\n \u003e out of sync. In the end, we just had to write lots of fixup\n \u003e routines and tools for operators to clean up the mess.\n \nI can unstand this, in our cloud, I was remembered the QUOTA error for Cidner project, the root reason is while create voume failed, the quota is occupied and recored in DB, the action is rollback but not cleanup the DB, so the quota is wrong while we use in the feature. (In Ocata version seems nova have the same issue, but I am not sure, that just a real case.) \n\nSo I think we should do this way, otherwise, if the resource account failed that we don\u0027t abort the DB save, we will also face data inconsistencies.\n\n \u003e My advice would be: if you\u0027re going to use placement for resource\n \u003e accounting coordination with other projects, do that, and don\u0027t let\n \u003e it get out of sync with you. (i.e. fail if it fails).\n\nAgree","accounts_in_message":[],"_revision_number":1},{"id":"f4e5647cb3815ef54c398b4081a8c1b5936a2de6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-21 16:09:12.000000000","message":"Patch Set 1:\n\nEmm.. I apologize for the incorrect indentation.","accounts_in_message":[],"_revision_number":1},{"id":"c921fe71d09b08e9ddc8ad36c4a1e27bdfe726ac","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-22 10:52:12.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"a2251bbbe54dd91704a3f13ae6d1440e7df9b64f","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-02-23 12:26:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nSo we have get agreement. \nLooks Good to me.","accounts_in_message":[],"_revision_number":1},{"id":"83eaf07db7c41924f7d9da5ca94d7455cf5b58e6","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-02-24 06:42:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Dan for fixing this and the advice!\n\n \u003e \u003e Xinran WANG:\n \u003e \u003e Do we agree Cyborg DB should not be independent?\n \u003e \u003e Should it never be updated if something wrong with placement\n \u003e \u003e update?\n \u003e \n \u003e No, Cyborg DB update should be independent with placement update.\n \u003e In current code, these 2 parts are in same functions, we should\n \u003e decouple them.\n\n \u003e My advice would be: if you\u0027re going to use placement for resource  \n \u003e accounting coordination with other projects, do that, and don\u0027t let  \n \u003e it get out of sync with you. (i.e. fail if it fails).\n\n+1. Agree that we should keep independent cyborg DB. And keep placement DB synced as well.","accounts_in_message":[],"_revision_number":1},{"id":"e73a38e6b0416b79d9b7144085e41fb0ac973da1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 06:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1deb9348e9e59d40206016c57da7364ff9af2d21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 07:05:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/393f4b10afbe4a24bf7446ec3186545e : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e133f8eb65154a3c9c5fb58957ca3260 : FAILURE in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3b133d90a2449a68f9406405aab73e2 : FAILURE in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce713f315934a82ae4c32bae1fade14 : SUCCESS in 5m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/35b593dc7cd1487e8774425696dad262 : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":1},{"id":"603037745bb3827a7dfa77aa3d075d82a89b34f7","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-24 14:15:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be57abc440051ae66dc8003ed3d6c86f860a7691","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-24 14:16:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c143d9fffb7212f62de38aaabbda6bc2ba68b78d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ab4aee4d3424c3f836a96010d84f5cc : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6def7d2582e44f96bd3b47edb95bf2c1 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a41fa92a9df54a7ca4aa3fd97a402c78 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51b9c2127656448383cd933bf1edd3a1 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90503753d1f74eda80caa0bc82e5e35c : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926a385aeba04683b7065558f14ceb9f : SUCCESS in 5m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/34095844f08a4e4cb6f1259864692c78 : SUCCESS in 35m 28s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1c6a99bf9348da8a803dd7f11299d7 : SUCCESS in 41m 14s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/52ed127495ce4770ba9d2bbf7c5be832 : FAILURE in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e20eeb2d4ac997611564d6e6fd290d18feea57a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:04:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3a646af1f0a4f9ad8869f6149eadd255b14a574","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:50:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"978dbcce7b26692ce71cb2ef99f7ee46ab1e5088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:50:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be79e77d94fd43ed8761ea10d74687e7 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a4c861c1d249609a61514327a594b3 : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5345b3344edb447785d715667e34f293 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40fddfdd89584432b8d30f0335dd307a : SUCCESS in 10m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f952606743b648a3b193901ce8b84252 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"64bbf86ebe1269df63110fc0edd2b28fbe0a4805","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:52:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa61ac9f2d8472db06564247315b6ae : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d279c22d1e3e1f053cd33bfb062fbaeb1f397472","revisions":{"d279c22d1e3e1f053cd33bfb062fbaeb1f397472":{"kind":"REWORK","_number":1,"created":"2020-02-19 18:49:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/708726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/26/708726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/708726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/708726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/708726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/26/708726/1"}}},"commit":{"parents":[{"commit":"e0ba01891fd479b42a95e76017b30d21b631430d","subject":"Merge \"Update gpu driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e0ba01891fd479b42a95e76017b30d21b631430d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-02-19 19:37:19.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-02-19 19:40:42.000000000","tz":-480},"subject":"Avoid creating a root provider when parent is not found","message":"Avoid creating a root provider when parent is not found\n\nBefore this change, when agent called to conductor to report_data(),\nif the parent provider was not found by hostname, we would log an error,\nand then continue to create the \"child\" provider with no parent. We should\nnever do this if we are supposed to have a parent. Cleanup from this\nsituation is also messy.\n\nThis makes us raise PlacementResourceProviderNotFound() in that case,\nwhich aborts the report and thus does not create the provider incorrectly.\nIt also makes the agent catch that exception and moves the log message\nto the agent where the actual problem is (i.e. likely misconfiguration).\n\nThe exception used here is actually defined incorrectly, having a message\nclass variable instead of _msg_fmt, which caused it to not render properly.\nThis fixes that along the way and adds tests for the new conductor and\nagent behaviors.\n\nCloses task: 38813\nCloses task: 38814\n\nChange-Id: Ied8ee91592eb0b4675f9c155e30a6c3a7df9b597\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d279c22d1e3e1f053cd33bfb062fbaeb1f397472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d279c22d1e3e1f053cd33bfb062fbaeb1f397472"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
