)]}'
{"id":"openstack%2Fcyborg~708770","triplet_id":"openstack%2Fcyborg~master~I5b6879e5ddb86659b8c7eb87b8d26cee33ed4754","project":"openstack/cyborg","branch":"master","topic":"multinode","hashtags":[],"change_id":"I5b6879e5ddb86659b8c7eb87b8d26cee33ed4754","subject":"move setting to devstack/settings","status":"MERGED","created":"2020-02-20 00:31:19.000000000","updated":"2020-03-03 07:16:55.000000000","submitted":"2020-03-03 07:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708770-1583219687826-cb471cbf","meta_rev_id":"8d052c981c99650e5f31f04b71a47e7fe21e32f4","_number":708770,"virtual_id_number":708770,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-02 15:16:20.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 07:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-03-03 06:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2020-03-02 15:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-02 15:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-03-03 06:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2020-03-02 15:25:43.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 15:11:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-03-02 15:24:30.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-03-02 15:25:43.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-03-03 06:34:49.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-03-03 07:14:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"040e9fcc95f91b4164b576dbd136e1b95bd60c1d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 00:31:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a38ab7f78170bac88897f51bfee5e4b9cea5fe19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 01:12:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce4cbf60cb71496fbb8573eeaec1c994 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1e21f84021405096580f36f46ffd3b : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea4aab505cf447aa55a368bbccfd50b : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5931a98b2524807962930e2499f627a : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e87037090575467a89a4e90e90c2006d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8185e6444f840c7984641f2bbb1f6ee : SUCCESS in 5m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8ca8c375935a4565bc36591d047bf82c : SUCCESS in 29m 24s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e7f448efd484b06868a6bbdf5a3dd83 : SUCCESS in 35m 20s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/2bb15a93ae954a40a88d1d6d13a25894 : FAILURE in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4298708dbf92f74c0ff31f351b1a63f428def0d3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-02-20 14:51:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\ntwo spelling nits, but otherwise I\u0027m sure this is good.","accounts_in_message":[],"_revision_number":1},{"id":"169a59dcd364be1c26f3be7910381abdf67d8acc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 15:51:53.000000000","message":"Patch Set 1: Workflow-1\n\nyep ill fix","accounts_in_message":[],"_revision_number":1},{"id":"4816558e31e6e5392ff0cd8dfd24db08be2e8e66","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-24 22:13:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"faa873f7e2e9570a1152e88c80199c2259c8b1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 23:10:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46c6f5da86ad4726ab90d0b4fbd79ab4 : FAILURE in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8de47f21ac34d618d3e66065cac765c : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e6817c436524248b4356800f0fac71d : FAILURE in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/227ac48ad19b41ea8a6fd0ead1a160e6 : FAILURE in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d34942a80e5742b592c58cdfe0488cda : FAILURE in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e48ca112f14a248aa9800433cca6c3 : SUCCESS in 5m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc2e4d03a2414bb596dce6ff0e09bc68 : SUCCESS in 29m 05s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6bc385c563c4dd38511a013c269ce1e : SUCCESS in 30m 39s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/823cc96e66f04ddea3700e8acf2e397e : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d7714cea705419825abf6c514379ebc967f2ce7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-24 23:19:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ae35a0dcbda15674530b63c6e599c0a4703532dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 00:07:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b8b7a2b0c174ea8927955769f29cc77 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/accc88faf18948848f03d8a35e455eaf : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d99b6f87ac34dee89057282ba8250c0 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/021655a42245480a992ea908ca10c644 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/332ddec2fcdc43e090b828392305dec0 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668794db6ae64068bdb2dd0cbfab4e21 : SUCCESS in 5m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f11ac7e9bf574e61b90bc9130c36660e : SUCCESS in 35m 37s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e16ec1612f23486e9bbb4eb4f49b84bf : SUCCESS in 42m 56s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/5e6dddbc76d4408ba081326a4a701c06 : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"567b14ac4aa253641a0ab59421af34b9daa5e419","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-02-25 14:51:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2a42a3696874d5a96876c91784675733bdf9f817","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-27 07:36:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThanks sean. soft -1 for nit.","accounts_in_message":[],"_revision_number":3},{"id":"935b1bac593bfa002e68a7748619dc44920898e5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-03-02 15:02:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1abdeb9c42a5bbbaf123213200da1e3f3940dfc9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-03-02 15:11:10.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nOne more nit, but we should probably just let this go :)","accounts_in_message":[],"_revision_number":4},{"id":"68c9c2f866656a8b944fad90e3590434b70b6442","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-03-02 15:16:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cb7467fb6c78bb0447ec63825c3131898de7da6c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-03-02 15:24:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"34e1c4e93a4971b53b19656ab0a6f80f6c2207c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 15:35:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddef7c5f2ec943f39f305aa624af6621 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3fbf5bbc9ab47f48cd4f8e0d4f9ab2b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b342426ce2749fa87e1b1eee8ed0145 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/051d83fda60a4399adc2bb57bd11957b : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d15b93ac19f1448d94232c0c6dd76278 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0fe5c2420f4b89b71e0ad47c75a9db : SUCCESS in 6m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c757f44f57a44dc899f4be5eeede23d3 : SUCCESS in 31m 10s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/400ff78a9e464b0692e4ff9a3e4eae3d : SUCCESS in 30m 28s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/6c007132cd1e422f8811e192ecb4774e : FAILURE in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6f4ed1144b84295090401f3dcf835142cf097a81","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-03-03 06:34:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nQuick update! Good patch.","accounts_in_message":[],"_revision_number":4},{"id":"908015dbbe5682ba30314cd2d53fb1521775eaba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 06:35:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5b654b33af199ead640821c2b647d47ded02307b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 07:14:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf63524d03548a4b7b4d798b75be398 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03999d4fa07e43d4a8276e696b359c11 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d774d27009346f9adbb49055397ed5b : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334acc8232a34bf991c0ab3277d9eda2 : SUCCESS in 6m 59s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/89d252154b8349b5ba41f808c9239c9c : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":4},{"id":"83096183c7d21af3f7129d5186642e1a1fcd35c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 07:14:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"69c10daeb13edab6209d9ceb2f1498de8aadfdd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 07:16:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d3ad6044ab4855bbc956e4a2d0a970 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"985b15287ee5a958ce9fb8861f35546e23d6bb66","revisions":{"958634a2ccedc6cc76e24d4da57c3e60bf84daa6":{"kind":"REWORK","_number":1,"created":"2020-02-20 00:31:19.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/70/708770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/70/708770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/70/708770/1"}}},"commit":{"parents":[{"commit":"94dd0ee63e4ebcd5ce49dc24f80e8513aa06f52b","subject":"add support for multi node deployments to fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/94dd0ee63e4ebcd5ce49dc24f80e8513aa06f52b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-19 20:55:23.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-19 23:56:10.000000000","tz":0},"subject":"move setting to devstack/settings","message":"move setting to devstack/settings\n\nAll user settable options should be stored the\ndevstack/setting file so they are defiend when it is\nsourced early in devstack to allow values to be shared\nbetween plugins if required.\n\nThis change moves the cyborg settings from\ndevstack/lib/cyborg to devstack/settings to conform to\nthe standard plugin interface.\n\nThe name of the folder where devstack clones the plugin\nis specified in the first argument ot the enable_plugin\nfunction invocation in the local.conf.\nThis change makes updates CYBORG_DIR to respect that\nand adds TODOs for other issues that will be adressed in\nfollow up patches.\n\nChange-Id: I5b6879e5ddb86659b8c7eb87b8d26cee33ed4754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/958634a2ccedc6cc76e24d4da57c3e60bf84daa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/958634a2ccedc6cc76e24d4da57c3e60bf84daa6"}]},"branch":"refs/heads/master"},"8cb1217e75d491dca7b78c2bf3b8f96b4106e64b":{"kind":"REWORK","_number":2,"created":"2020-02-24 22:13:54.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/70/708770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/70/708770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/70/708770/2"}}},"commit":{"parents":[{"commit":"c21373ed612f1e725ffb27ecd3d066024d6e0d41","subject":"add support for multi node deployments to fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c21373ed612f1e725ffb27ecd3d066024d6e0d41"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-19 20:55:23.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-24 21:46:05.000000000","tz":0},"subject":"move setting to devstack/settings","message":"move setting to devstack/settings\n\nAll user settable options should be stored the\ndevstack/setting file so they are defiend when it is\nsourced early in devstack to allow values to be shared\nbetween plugins if required.\n\nThis change moves the cyborg settings from\ndevstack/lib/cyborg to devstack/settings to conform to\nthe standard plugin interface.\n\nThe name of the folder where devstack clones the plugin\nis specified in the first argument ot the enable_plugin\nfunction invocation in the local.conf.\nThis change makes updates CYBORG_DIR to respect that\nand adds TODOs for other issues that will be adressed in\nfollow up patches.\n\nChange-Id: I5b6879e5ddb86659b8c7eb87b8d26cee33ed4754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8cb1217e75d491dca7b78c2bf3b8f96b4106e64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8cb1217e75d491dca7b78c2bf3b8f96b4106e64b"}]},"branch":"refs/heads/master"},"c9d1845f26bde67bf2db931ed979a83bfd1a0d83":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-24 23:19:00.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/70/708770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/70/708770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/70/708770/3"}}},"commit":{"parents":[{"commit":"eb12f68421bac1c0dd410ea2f0bbb828797dfcd3","subject":"add support for multi node deployments to fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eb12f68421bac1c0dd410ea2f0bbb828797dfcd3"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-19 20:55:23.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-24 23:16:13.000000000","tz":0},"subject":"move setting to devstack/settings","message":"move setting to devstack/settings\n\nAll user settable options should be stored the\ndevstack/setting file so they are defiend when it is\nsourced early in devstack to allow values to be shared\nbetween plugins if required.\n\nThis change moves the cyborg settings from\ndevstack/lib/cyborg to devstack/settings to conform to\nthe standard plugin interface.\n\nThe name of the folder where devstack clones the plugin\nis specified in the first argument ot the enable_plugin\nfunction invocation in the local.conf.\nThis change makes updates CYBORG_DIR to respect that\nand adds TODOs for other issues that will be adressed in\nfollow up patches.\n\nChange-Id: I5b6879e5ddb86659b8c7eb87b8d26cee33ed4754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9d1845f26bde67bf2db931ed979a83bfd1a0d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9d1845f26bde67bf2db931ed979a83bfd1a0d83"}]},"branch":"refs/heads/master"},"985b15287ee5a958ce9fb8861f35546e23d6bb66":{"kind":"REWORK","_number":4,"created":"2020-03-02 15:02:19.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/70/708770/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/70/708770/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/708770/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/70/708770/4"}}},"commit":{"parents":[{"commit":"eb12f68421bac1c0dd410ea2f0bbb828797dfcd3","subject":"add support for multi node deployments to fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eb12f68421bac1c0dd410ea2f0bbb828797dfcd3"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-02-19 20:55:23.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-03-02 15:02:17.000000000","tz":0},"subject":"move setting to devstack/settings","message":"move setting to devstack/settings\n\nAll user settable options should be stored the\ndevstack/setting file so they are defiend when it is\nsourced early in devstack to allow values to be shared\nbetween plugins if required.\n\nThis change moves the cyborg settings from\ndevstack/lib/cyborg to devstack/settings to conform to\nthe standard plugin interface.\n\nThe name of the folder where devstack clones the plugin\nis specified in the first argument ot the enable_plugin\nfunction invocation in the local.conf.\nThis change makes updates CYBORG_DIR to respect that\nand adds TODOs for other issues that will be adressed in\nfollow up patches.\n\nChange-Id: I5b6879e5ddb86659b8c7eb87b8d26cee33ed4754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/985b15287ee5a958ce9fb8861f35546e23d6bb66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/985b15287ee5a958ce9fb8861f35546e23d6bb66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
