)]}'
{"id":"openstack%2Fcyborg~716916","triplet_id":"openstack%2Fcyborg~master~If9cd7eca3454d9e1d62a0a1753e157adbf965e8d","project":"openstack/cyborg","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"If9cd7eca3454d9e1d62a0a1753e157adbf965e8d","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-02 10:21:00.000000000","updated":"2020-06-19 08:29:17.000000000","submitted":"2020-06-19 08:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716916-1592555253350-4158ad97","meta_rev_id":"2e9066e0b04b3a0441a9e2b767693340830d0418","_number":716916,"virtual_id_number":716916,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"date":"2020-06-17 00:54:51.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 08:27:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":2,"date":"2020-06-17 01:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 07:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 07:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 12:28:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-16 03:56:49.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-06-16 03:56:54.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2020-06-17 01:26:31.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-06-19 07:44:22.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-06-19 08:27:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b788e8f1335156072b5073fca9744223d89f131","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-02 10:21:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9662b821a031ec90ce57cceb5bf15936743150bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 10:57:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20197ff8cde64f14b2e76b1d63ae4b35 : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a624fc6d3297406d8a1bc15d5319b546 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/654087ab2bf14e7f8ebfe1fceee11d79 : FAILURE in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/023aa62c2eef4dc5a459ec1fc9548083 : FAILURE in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baac5c26af314b8ea15895a37a84f323 : FAILURE in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2bab5ba08b4abdad611540ab69b410 : SUCCESS in 6m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/120021dabb8b455a9fd696c6779239ac : SUCCESS in 31m 24s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70690ebe1d954b2595180c0f46d6ab13 : SUCCESS in 30m 52s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/0c09cf9172d64c7aa80ced27f178e4c5 : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5de84b14466ddbeec4a6b37b444d3209fd52c498","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-03 09:23:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a40f524f29a9ffbade5a5082b7286f30065255f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-03 10:06:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e2a28fed019a19c250603420840d42aeb9b2796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:50:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc674ab8bb3f48d480ceae1a255bf5f1 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9584b1ff3c34b54add43324bb1cbbc1 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6a5ebf20df9467cae1369fc202ece47 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f7beb4538ec4896a3eff37225bc3900 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f5279f33b64ed1bd61145b38fb458f : FAILURE in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5f27d4a41b4010b50334bcd8a65a8d : SUCCESS in 9m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9a55bf8a1b2f4d59a28fc13649aab786 : SUCCESS in 31m 50s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d00fa49ee554f56a6490746a3e8f96b : SUCCESS in 40m 15s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/b17fc3308122470994e94f35ccdaa01e : FAILURE in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c55f42a6d1df00d78ea85554dd1089664c180c33","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-15 12:28:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1aeec44c817b0fc69025d8495488bc60c7f0e7fe","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-26 08:17:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"69d3250697cc3cac146fb46f7e6e58094968c3a1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-26 08:18:10.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks Sean","accounts_in_message":[],"_revision_number":2},{"id":"626898f71128615118b2a7d2e8e88571f5fe7b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 08:53:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e79def58e98435898e6574e96a29651 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c839a9c3a71f415eba7fc03d77c244a8 : SUCCESS in 4m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d93a2c4ca1024901b27440a4aa6bb02f : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac959b2cf37d4b9e8268e1e298ed3e12 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc2e58241e64337b91ec9227258e785 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef0f6e163b6c47c7995e2305b33fcb38 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71cbaac462fb463982791d28451e9b9a : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d5fa36c1f1342ec9df3474544c73f36 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89add785dd7d4040a927628701520bc7 : SUCCESS in 5m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/975e0d68a0854cf0b18d193d8cbc444b : SUCCESS in 31m 12s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea70e8e5b075400693aa7dae9ed4aa7c : SUCCESS in 29m 02s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/d215579be83f48dbb6b3fe9b70fcba1a : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"99552b3a5db9a6162698379faaa0f0cd3dffc88c","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-06-16 03:56:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"00dc2894d8077803a7295a43dd4ab9580764be1d","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-06-16 04:15:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nThanks, Brin! requirements.txt also needs to remove mock. The rest LGTM.","accounts_in_message":[],"_revision_number":4},{"id":"a103eef10bd0debe5603bdb74841219758d5627f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 04:27:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/770d84a320374540a86882c8643f5ffd : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d2484ad67c74f4cb936a8dbbe611844 : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1af6154fc82f4a2e9f1d4e97b78d66f5 : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baad396f7e454dfb82244097ce2f9222 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9490e063df6e463cb74f9ae373bfea0a : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4584b31a4e0b48aa8e5ff7d713c14b38 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dcbf8a4cf0849508d5150dee8a42930 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93acb9ef136414ead9073b7e184603f : SUCCESS in 6m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/496ca4be42f04a848c10847903dd677e : FAILURE in 21m 41s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/761cffdc6ea842478cc8bd91a84cd6e3 : FAILURE in 26m 22s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/a8253afebb3440f89759f9fb8a3169ab : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"14fccdc075e6ebe9dbf9bf06ae9b3e89e393a477","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-06-16 07:24:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"4a3ed5178cc6c265697cac00ab68a3a52721bf52","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-06-16 07:25:37.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThanks yumeng.","accounts_in_message":[],"_revision_number":5},{"id":"7848214de7e9167f1430e68c9f9b43502a235eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 08:26:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cfb296b188684ac99fb75063716f5c0e : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfef168fab034835a68daf0f8433c46e : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b30c7a73c0bb46e8a0c150df1d1531a6 : FAILURE in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34fed91f8f9d471584ee2b5706a922be : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f91b74713d4895adc436271c41cdb1 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b73a1090ccd447f92360fd618e2717a : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3917399ccb80467eb9cabe110711b817 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b4656119a241e495a2b35909071800 : SUCCESS in 9m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9349fca7fb50400ba2211c7008510aa9 : FAILURE in 27m 48s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b166e92f80749ccabe09eb02927f930 : FAILURE in 21m 01s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/26db34f2563448d1975a81d7f9e885cc : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":5},{"id":"8d0f5dbcadca1b224dac9a2a175474b97e2fa7b4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-06-17 00:54:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9e86f37aaa0cb9897522c4341189de629a892079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 01:25:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/14537647f48a49bdb072bf430cb66306 : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b40ab8a8a874658b6d67cbc2a6a3377 : SUCCESS in 6m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b8514160341340fa93e6cd507b854c7c : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66bf4a16f1fc487692a21adfab38af21 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df09fda2823f4dacb6c953291905cee0 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57093ae7ea0a493fbbd6967b08049ad9 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b53b40cae5745e693cb2204d7839327 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a49e6de43241969001f69aebd99efd : SUCCESS in 6m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f7cc28f45c724ec09293014fc8e30976 : SUCCESS in 28m 59s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5e03d30f44649eea0bd3f1f7c7afec5 : SUCCESS in 25m 33s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/354acbb5cc344f299ebb919ea1223208 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":5},{"id":"d24aa8455624681160690952a73b3a86610c439e","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-06-17 01:26:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2441988c57e258ee6bc4d1c4e519aeac530ede99","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-06-19 07:44:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"76341d20c2bd7a0aba2a5e99923730bb58be2bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:44:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cd5e9738a5918ffea7f01f8ae13cc4114e85be86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:27:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"feaf32cd73ab206d651b1b2b7f76f6c2d3ef7782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:27:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/43dbcc9e529d46ab9d62f9ec738ef584 : SUCCESS in 7m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ed06ac21c6a48c9868aa288d12dd045 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00ddcb517e9e4129bfa572538ecc04e6 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d3821444724765ac13bdbae6b373a7 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c59c0fe4a1042089e79af1c6b8c4f02 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b91ed04e0a540009b7c6a2e052eefe3 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d9504a13ae499dbd9cf42e7cf51cbe : SUCCESS in 11m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/efb201bdf9f64225849867cdae3a9bdd : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":5},{"id":"370336e78bbc36a26900992e65ae16908e6a46bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:29:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/16e99fa57463462ebd42801a074460f3 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fd40c16798246a9808324db7c47b791 : SUCCESS in 1m 21s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d58242adb248b4aab4fa72e35798bd : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6b0e1f8d12fdfd50c6bcde507bf603a2bf856a18","revisions":{"4cfd176ffd6451a514e5ca5c8c29d3045330e698":{"kind":"REWORK","_number":1,"created":"2020-04-02 10:21:00.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/16/716916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/16/716916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/16/716916/1"}}},"commit":{"parents":[{"commit":"6595fd7e3376cb95a9be67eac224575c8c032e01","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6595fd7e3376cb95a9be67eac224575c8c032e01"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-02 10:17:34.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-02 10:17:37.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib. Most of this\nis autogenerated, as described below, but there are three manual changes\nnecessary.\n\nChange-Id: If9cd7eca3454d9e1d62a0a1753e157adbf965e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4cfd176ffd6451a514e5ca5c8c29d3045330e698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4cfd176ffd6451a514e5ca5c8c29d3045330e698"}]},"branch":"refs/heads/master"},"368698abc6a5b3fc638a2cf653bffefef9263dc7":{"kind":"REWORK","_number":2,"created":"2020-04-03 10:06:05.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/16/716916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/16/716916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/16/716916/2"}}},"commit":{"parents":[{"commit":"6595fd7e3376cb95a9be67eac224575c8c032e01","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6595fd7e3376cb95a9be67eac224575c8c032e01"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-02 10:17:34.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-03 10:03:33.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib. Most of this\nis autogenerated, as described below, but there are three manual changes\nnecessary.\n\nChange-Id: If9cd7eca3454d9e1d62a0a1753e157adbf965e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/368698abc6a5b3fc638a2cf653bffefef9263dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/368698abc6a5b3fc638a2cf653bffefef9263dc7"}]},"branch":"refs/heads/master"},"efa46e931bb8d30fd2c498798e76fcacc2d7b2cd":{"kind":"REWORK","_number":3,"created":"2020-04-26 08:17:38.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/16/716916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/16/716916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/16/716916/3"}}},"commit":{"parents":[{"commit":"6595fd7e3376cb95a9be67eac224575c8c032e01","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6595fd7e3376cb95a9be67eac224575c8c032e01"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-02 10:17:34.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-04-26 08:17:29.000000000","tz":0},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib. Most of this\nis autogenerated, as described below, but there are three manual changes\nnecessary.\n\nChange-Id: If9cd7eca3454d9e1d62a0a1753e157adbf965e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/efa46e931bb8d30fd2c498798e76fcacc2d7b2cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/efa46e931bb8d30fd2c498798e76fcacc2d7b2cd"}]},"branch":"refs/heads/master"},"3b4f8e4212ee81f6110d1ac37a4ba716316cb2d5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-16 03:56:27.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/16/716916/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/16/716916/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/16/716916/4"}}},"commit":{"parents":[{"commit":"8ed9135de989c4354503d8386425ecfb2d8f87d9","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8ed9135de989c4354503d8386425ecfb2d8f87d9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-02 10:17:34.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-06-16 03:56:27.000000000","tz":0},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib. Most of this\nis autogenerated, as described below, but there are three manual changes\nnecessary.\n\nChange-Id: If9cd7eca3454d9e1d62a0a1753e157adbf965e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3b4f8e4212ee81f6110d1ac37a4ba716316cb2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3b4f8e4212ee81f6110d1ac37a4ba716316cb2d5"}]},"branch":"refs/heads/master"},"6b0e1f8d12fdfd50c6bcde507bf603a2bf856a18":{"kind":"REWORK","_number":5,"created":"2020-06-16 07:24:50.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/16/716916/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/16/716916/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/16/716916/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/16/716916/5"}}},"commit":{"parents":[{"commit":"8ed9135de989c4354503d8386425ecfb2d8f87d9","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8ed9135de989c4354503d8386425ecfb2d8f87d9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-02 10:17:34.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-06-16 07:24:42.000000000","tz":0},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib. Most of this\nis autogenerated, as described below, but there are three manual changes\nnecessary.\n\nChange-Id: If9cd7eca3454d9e1d62a0a1753e157adbf965e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6b0e1f8d12fdfd50c6bcde507bf603a2bf856a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6b0e1f8d12fdfd50c6bcde507bf603a2bf856a18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
