)]}'
{"id":"openstack%2Fcyborg~727483","triplet_id":"openstack%2Fcyborg~master~Ia12b837b81faec9288bb649c4d544f7eadf5c8b3","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"Ia12b837b81faec9288bb649c4d544f7eadf5c8b3","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-13 00:23:13.000000000","updated":"2020-05-18 18:45:36.000000000","submitted":"2020-05-14 03:36:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727483-1589427407905-d3101825","meta_rev_id":"a3e587a025690799953b2be315e06a14c9388ab1","_number":727483,"virtual_id_number":727483,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 03:36:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":2,"date":"2020-05-14 03:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2020-05-13 03:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":1,"date":"2020-05-14 03:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 03:20:11.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-05-14 03:28:36.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-05-14 03:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8233f9a2e1086c2168a5da37ec71573c816dfb33","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 00:23:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"485995764ff18b00d466c2c40680cfb75a10a92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 00:38:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/60acb47aa5574bedb67a84234e3bdb31 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73c9a6d45a224b28a564f600a79ddd8e : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3185d7a101f543748b6f8bd1c9d5a58a : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c5b20c8ef9c46288aa491c5f6d56361 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2e50fd81e64f80a7b8b70bc781b487 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3af8d39df744edf9c6d38de27b4b8da : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33af82b2e3364fd79c64502d03957f6d : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc9bab2c02e47baae344215085aae9a : SUCCESS in 6m 32s\n\nWarning:\n  Comments left for invalid file cyborg/tests/unit/accelerator/drivers/fpga/intel/prepare_test_data.py","accounts_in_message":[],"_revision_number":1},{"id":"a598bffbedeeb7b0162d4b9aaad41e6ec211dfe1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-13 02:05:10.000000000","message":"Patch Set 1: Code-Review-1\n\nyou should fix the related issue.","accounts_in_message":[],"_revision_number":1},{"id":"84c22ae4608b64150496e9b7f2df33558cbce92c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 02:59:05.000000000","message":"Patch Set 1:\n\nhacking and flake8 version are good but pycodestyl is 2.6.0 which caused the issue","accounts_in_message":[],"_revision_number":1},{"id":"d96c6bb6ea0f5bb775a31d5ec93512d5c5aa1697","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 02:59:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a54520c6f4e8ba51d209424ff3e97341d37125de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:14:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8d03bd4fb65f432e9344aff2fcf7774a : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a6ab7c01e394d9384a102662ed1ac3f : SUCCESS in 7m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7adb73c136a5437ab77d0b9cfec6cc73 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/598e7095f6674fd4ac0b1db3cc360204 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30321e10d2d48ba8e8054f73d8e7830 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf24a53389744d2afaef549ff887241 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2210782d3ab848d7a44d70b9a8b8146d : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c035f33644f14c388e6feeea47f1e8ea : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f2b4fa437d3e215495d11bf3ba43381a515b1a03","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-13 03:20:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f45c8e1779868b5af2731588f42dafcd53d8575c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-05-14 03:28:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2fdf59f50319ec314cb2cf0b6b4c905949e8fa06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 03:28:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"956abcd228ef380fb29ef3f35c40a92948a8d93e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 03:36:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"452b27cf198c2d766c347df445e7343021b9a176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 03:36:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/24408200454742e5ac43546a530d277b : SUCCESS in 4m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9ad0507bcebd4b688a9b4b697901ae18 : SUCCESS in 2m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fec808731b80406fa23a44bdf7a0b3a0 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770e9060b0094831993c4c4d7afaaef7 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9d513f71aec4324aecf96dde6879321 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f0689ebffe645d7b813966236f0efff : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db61c20ce42d4a2d9893a196961011b5 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a84b044063bfbdf7b5c401f2a6797a96b99d9857","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 03:38:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2a3c286bc0fa4a8290a75689602e2d6d : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91dd1d7a1d6545dba2ffc4de6344d3a5 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffefba7099904a92b976f0960e06ea00 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"a3e587a025690799953b2be315e06a14c9388ab1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-18 18:45:36.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit ef17dbd1b6a1db8d3914db7fb3721e677b12779b","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4c941dbc8199e105fb65055f24340c07badce452","revisions":{"4f9d83a90e7c1d8ce7fa3cc747ea4bb27f5ded40":{"kind":"REWORK","_number":1,"created":"2020-05-13 00:23:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/83/727483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/83/727483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/83/727483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/83/727483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/83/727483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/83/727483/1"}}},"commit":{"parents":[{"commit":"0a03e91c6c71af1b21bfc09a970ed492eb1f92e3","subject":"Merge \"Fix fake attribute old interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0a03e91c6c71af1b21bfc09a970ed492eb1f92e3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-12 22:52:13.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 00:23:12.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: Ia12b837b81faec9288bb649c4d544f7eadf5c8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4f9d83a90e7c1d8ce7fa3cc747ea4bb27f5ded40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4f9d83a90e7c1d8ce7fa3cc747ea4bb27f5ded40"}]},"branch":"refs/heads/master"},"4c941dbc8199e105fb65055f24340c07badce452":{"kind":"REWORK","_number":2,"created":"2020-05-13 02:59:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/83/727483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/83/727483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/83/727483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/83/727483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/83/727483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/83/727483/2"}}},"commit":{"parents":[{"commit":"0a03e91c6c71af1b21bfc09a970ed492eb1f92e3","subject":"Merge \"Fix fake attribute old interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0a03e91c6c71af1b21bfc09a970ed492eb1f92e3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-12 22:52:13.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:59:19.000000000","tz":0},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: Ia12b837b81faec9288bb649c4d544f7eadf5c8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4c941dbc8199e105fb65055f24340c07badce452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4c941dbc8199e105fb65055f24340c07badce452"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
