)]}'
{"id":"openstack%2Fcyborg~729852","triplet_id":"openstack%2Fcyborg~master~I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63","project":"openstack/cyborg","branch":"master","topic":"inspur-fpga-driver","hashtags":[],"change_id":"I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63","subject":"Add Inspur FPGA driver","status":"MERGED","created":"2020-05-21 07:05:22.000000000","updated":"2020-09-10 06:14:02.000000000","submitted":"2020-09-10 06:12:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729852-1599718323984-2f639b6f","meta_rev_id":"de5b8d91f30934600b89a734187dbe78be2215d3","_number":729852,"virtual_id_number":729852,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 06:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2020-09-10 05:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-09-10 02:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2020-09-10 05:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 07:15:16.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-08-27 01:36:57.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-08-27 07:43:19.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-09-10 02:49:26.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-09-10 05:36:11.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-09-10 06:12:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a695e74b2186cbf76c979a7711122756e6a9770b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-21 07:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c642f293eaff1f655bd8b8f728f9568e568fc02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 07:43:50.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d18cbce31458440899af77a8c117fec9 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b7422c6da094c46b1af7c4d8c90a693 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab97a4bff66a47e2856bb10bd54aeeae : FAILURE in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f6988febe3c42b0b447f20eca154f72 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72ede0c69feb45e49528762d8295a8ad : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdea75996e2e404fa66b081059e034b1 : SUCCESS in 7m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/80b35be8aabe44e3aac87c766d0bbf50 : SUCCESS in 33m 16s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/819c42be488d4e099f5abd6372e2edd1 : SUCCESS in 36m 51s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/749dfe72d1b4428f8b339853fb0c58c6 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"eebb8a9af50defa1149918685525695b5298caa1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-21 08:39:04.000000000","message":"Patch Set 1: Code-Review-1\n\nThis need a spec to explain this fpga drive, like Intel QAT driver https://review.opendev.org/#/c/728014/","accounts_in_message":[],"_revision_number":1},{"id":"6fa88f7e822fa5b07ab519c6a8fa2a3797f3df64","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-26 11:01:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8c20f78d25be801555b75abcf2fc1b62f26e50d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 11:31:38.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10c1fe7ad9b841698f2c1ab4da274ed5 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaea298b014f49d0ac0761b349c67e57 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15da05ae630c42a4ad85be97a12766df : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92de2c72139f4c46bf59fe923210aff9 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50013de997a445419d0c7513e8eb3c4d : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca0f80dea3b421f9d84d63be7bfc083 : SUCCESS in 5m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b609297360364ce7bef0995b113f65be : SUCCESS in 26m 31s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78725b40ff5f44acba2d5149c0f57706 : SUCCESS in 28m 59s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/d892fffbbffe45b48e12267996e42b6d : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b16f0a6acc61516a408b1646a01d613c9116b47d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-27 01:28:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3a7b058b20a2c87c3b8d694fd96456aa2b215107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 01:56:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b3fc96c9a7d42168805a3052c5dd4fd : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9505133fde14fd7afd7ac96e1b68fc1 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fdf52c914684d3fbdb7b92085f5ad09 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0788391a631f4140894b168e62faede0 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d96f78285f4d1b8e518dbdc0b0cfe1 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b892861215c47bfa7ea0e9cd0453e95 : SUCCESS in 5m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/af49ee2d8a78496c974632902119d440 : SUCCESS in 24m 12s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2055ec02609f469f82a9a303e488a548 : SUCCESS in 27m 23s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/2f58574b1cdd40538fffdcdc8c1885f0 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"4d3132d8b215ef7d763f40d7880a486ea0964277","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-21 10:01:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1e0babdf9d1d7df59968eeb8de213e9bda7ff9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:42:26.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6fb8bef7a7304dbfa41d6240fcf6c73b : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2085a6079be4233b3e4bea02e204d51 : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c784f8eb817b4f619c897d49997b0a0f : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f6191ea81114dc08401050d961e8cb4 : FAILURE in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deadcc403b324a2ca642abb749e74a6a : FAILURE in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af39c3d6b094f62a9b0f1b989b0977e : SUCCESS in 6m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/183c4a668c094c0389db5f50ccb9df4b : SUCCESS in 32m 25s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35fdc76e57364c4188f045ff6b799393 : SUCCESS in 32m 07s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/dba651510e4341098d0a80f78b7d77a2 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":4},{"id":"3984bc894d9167a76a94b5abfcd240349a3ad522","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-21 11:33:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"19b35f977ea91b68afbf5565f762b18a3920e0d0","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-21 11:37:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"694d85d6a56fd940677fb0c0db452c91762dcf39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:22:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9aea09057c444c29b1d1a74ee86fcdbf : SUCCESS in 4m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1561c768a2c4feba9c872f4a1a96c44 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4087f617254f17b71f7c6e57e105e9 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c2849d206da41fab4acbc31e6703c25 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/648df87778c049d8b62bb96907ca99f3 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d05dbdaf1bf4e1eadee4b8265714472 : SUCCESS in 7m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2d88dceb5f2c4e1789cbffb6a89199b3 : SUCCESS in 29m 42s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72555a2f6cf4440685a0c0d8c757a37e : SUCCESS in 31m 02s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/10f2c3c7b00c4d99b5be39ae23a64e80 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":6},{"id":"2188ab5955e140a6ac5af8ba7893e1a74b795ddb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-25 02:17:46.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM, a question where we can save the temporary report result, in the core dir? or the test dir?","accounts_in_message":[],"_revision_number":6},{"id":"1e867765d75585b11948e2def94d30b01dd0b1f4","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-08-27 01:36:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b55c1f3d5624aa4ab667af1beb984750242ac3be","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-27 05:57:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a95b8c7bd9c4f6fe474c7f3668601e00d72ab52e","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-08-27 07:43:19.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nHi wenping, overall looks good to me. pls check my comments inline.","accounts_in_message":[],"_revision_number":6},{"id":"4548e20ed75c11da0d36cd9d90b13ac6d994a1b7","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-01 02:41:35.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"d04b9270b7c6273e356393b12d976c9171ccfb4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 03:09:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8672083bd6784db79fe34c30486975a3 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6443be3e3c44f786c3453e06ad16dd : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1295368f738848a5a9a9fc82cfdf94c0 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bec754d038b4df7abe8f01279a55e4a : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb8578cd64e4439b5d700fa14b9b423 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7614898f694a9a981aa37289f01904 : SUCCESS in 7m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8046ec23341648568d75d2ecf0c289d9 : SUCCESS in 26m 50s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2073e8dc98f43bc865465ba436f36e8 : SUCCESS in 24m 17s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/73fe4fd7926145a78468eb71ead65dee : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":7},{"id":"2c9f22720047ff748f9df25edbfe527839bb3757","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-01 06:39:04.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nThanks Yumeng, plz check again.","accounts_in_message":[],"_revision_number":6},{"id":"f5dc525eda439fa862949b8f22e2f57520fabce9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 01:02:26.000000000","message":"Patch Set 7: Code-Review+2\n\nstill LGTM, how to show the lengency \u0027deployable_name\u0027 can be done by follow up.","accounts_in_message":[],"_revision_number":7},{"id":"3c1a95163034aca9eb6c27b33993d868ec4ab7e9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 01:04:18.000000000","message":"Patch Set 7:\n\nYou also need to update the driver matrix by follow-up.\n\nhttps://opendev.org/openstack/cyborg/src/branch/master/doc/source/reference/support-matrix.rst","accounts_in_message":[],"_revision_number":7},{"id":"9146bfff3b04ce008e80970d8f7ba34e69818244","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 01:07:11.000000000","message":"Patch Set 7:\n\n\u003e still LGTM, how to show the lengency \u0027deployable_name\u0027 can be done\n \u003e by follow up.\n\nsorry for the spell error. changing: s/lengency/legacy/g","accounts_in_message":[],"_revision_number":7},{"id":"7728ad3de7cd6080e8495692ebe63b74b70f27f5","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-05 02:38:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ceff8b9795d23af8d35ff878af16edf3cb15aa20","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-08 09:56:59.000000000","message":"Patch Set 7:\n\nOne question here: Can this FPGA be virtualized, I did not see any VF management in this driver.","accounts_in_message":[],"_revision_number":7},{"id":"cfb893cda966e6796133821e33f175f1208f97a6","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-08 09:59:59.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nSoft -1 here, please see my comment inline.","accounts_in_message":[],"_revision_number":7},{"id":"787228576d0c1a989009c442d3db12d56fcf42fb","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-08 12:18:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"81ac07ddc7f2eb221cb9764110ed621aea129ec6","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-08 12:20:14.000000000","message":"Patch Set 7:\n\n(1 comment)\n\n\u003e One question here: Can this FPGA be virtualized, I did not see any\n \u003e VF management in this driver.\n\nThanks Xinran. This card cant be virtualized now.","accounts_in_message":[],"_revision_number":7},{"id":"ce74bd77362ff70e303ebcb20ab2d92228a2b038","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-08 12:40:00.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"a327ae0f80ab70e105f362dc7cf574a55f0478be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:13:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aba0fbd575a444fc94c8cef3e7fedca7 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0bd6e10f3a64881bba9638fef7ac801 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cafc2df25a2a47e4a7262dcb0c15c8ad : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f86e88c3759d416f89bcb91645ffa66e : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a83d7b917ef48a5baf6674184ff4e2a : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77a1b6367e545cda631a6a36f168e16 : SUCCESS in 6m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ff6ca68e88c14fb58a826639c53975a9 : SUCCESS in 30m 41s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a1a6df6b2b24a5a93582bc21752311e : SUCCESS in 24m 28s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/9b1f837a65b145ddbed21d78041bbe9b : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":9},{"id":"196cc305fd110d69251383434c8ac7fe0c8b5e2b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-10 02:49:26.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"15bd489fd99d61dce10b4aefd0d621c67942f652","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-10 05:36:11.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nLTGM, thanks.","accounts_in_message":[],"_revision_number":9},{"id":"464af26c0bd1e5a0bae9b494a57a2e28574ea417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 05:36:23.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"17e157c3958955423ecdfb946de2e7ccf18aaaaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:12:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"59734a16ff06d7547de3e5ae2adc15e20248ff05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:12:03.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d42e48e6152e40938ece7d15ac7bcd20 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d877dd81d34f9f8fb999612d74308e : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b953f60bf753491a95f3d6ed7bc4ade7 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c12cbff59a644cff877f1c26c3e5bc7f : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5382e6c5a38496997fbca4c889dc055 : SUCCESS in 13m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8acdd2e99824441bb2795ab9024c5335 : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":9},{"id":"a66f60d9a80b45fcaec443486d5cf7187027e7df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:14:02.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c1ca657dd3421e92d9b894a9a0fa09 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3108d28eb3b50999d662765d5d8b90df760ff580","revisions":{"55d0b7fa63e24c9019ca2ba692e1776a2da16e58":{"kind":"REWORK","_number":1,"created":"2020-05-21 07:05:22.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/1"}}},"commit":{"parents":[{"commit":"0a03e91c6c71af1b21bfc09a970ed492eb1f92e3","subject":"Merge \"Fix fake attribute old interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0a03e91c6c71af1b21bfc09a970ed492eb1f92e3"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/55d0b7fa63e24c9019ca2ba692e1776a2da16e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/55d0b7fa63e24c9019ca2ba692e1776a2da16e58"}]},"branch":"refs/heads/master"},"c943f3cdab534697f0bdcb8aaa3d18bba6372b90":{"kind":"REWORK","_number":2,"created":"2020-05-26 11:01:28.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/2"}}},"commit":{"parents":[{"commit":"0a03e91c6c71af1b21bfc09a970ed492eb1f92e3","subject":"Merge \"Fix fake attribute old interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0a03e91c6c71af1b21bfc09a970ed492eb1f92e3"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-26 11:00:37.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nAdd unit test\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c943f3cdab534697f0bdcb8aaa3d18bba6372b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c943f3cdab534697f0bdcb8aaa3d18bba6372b90"}]},"branch":"refs/heads/master"},"afee3f0cd8f362a9e1023633bf0c2545c325bbed":{"kind":"REWORK","_number":3,"created":"2020-05-27 01:28:04.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/3"}}},"commit":{"parents":[{"commit":"0a03e91c6c71af1b21bfc09a970ed492eb1f92e3","subject":"Merge \"Fix fake attribute old interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0a03e91c6c71af1b21bfc09a970ed492eb1f92e3"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-27 01:27:45.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nAdd unit test\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/afee3f0cd8f362a9e1023633bf0c2545c325bbed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/afee3f0cd8f362a9e1023633bf0c2545c325bbed"}]},"branch":"refs/heads/master"},"7bf5e7d1c03435a9c3c93cfdf3d79137abd8009a":{"kind":"REWORK","_number":4,"created":"2020-08-21 10:01:53.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/4"}}},"commit":{"parents":[{"commit":"0a03e91c6c71af1b21bfc09a970ed492eb1f92e3","subject":"Merge \"Fix fake attribute old interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0a03e91c6c71af1b21bfc09a970ed492eb1f92e3"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-08-21 09:48:59.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nPlease see the test report in\nhttps://etherpad.opendev.org/p/inspur-fpga-driver-test-report.\nThis patch implemented Inspur FPGA driver in Cyborg. The spec is already\nmerged.\nPlease see:\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/victoria/approved/inspur-fpga-driver-proposal.html\n\nStory: 2007772\nTask: 39997\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7bf5e7d1c03435a9c3c93cfdf3d79137abd8009a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7bf5e7d1c03435a9c3c93cfdf3d79137abd8009a"}]},"branch":"refs/heads/master"},"5e0791e2ceeb5a414f377bf71b6dd4db25919d69":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-21 11:33:42.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/5"}}},"commit":{"parents":[{"commit":"40259a25ecc00520c19e88789b77306d8db1b272","subject":"Revert \"Add new default policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40259a25ecc00520c19e88789b77306d8db1b272"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2020-08-21 11:33:42.000000000","tz":0},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nPlease see the test report in\nhttps://etherpad.opendev.org/p/inspur-fpga-driver-test-report.\nThis patch implemented Inspur FPGA driver in Cyborg. The spec is already\nmerged.\nPlease see:\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/victoria/approved/inspur-fpga-driver-proposal.html\n\nStory: 2007772\nTask: 39997\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5e0791e2ceeb5a414f377bf71b6dd4db25919d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5e0791e2ceeb5a414f377bf71b6dd4db25919d69"}]},"branch":"refs/heads/master"},"a1ab5986d8bd264e97a4bbf19cfe9b76124339dd":{"kind":"REWORK","_number":6,"created":"2020-08-21 11:37:25.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/6"}}},"commit":{"parents":[{"commit":"40259a25ecc00520c19e88789b77306d8db1b272","subject":"Revert \"Add new default policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40259a25ecc00520c19e88789b77306d8db1b272"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-08-21 11:37:03.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nPlease see the test report in\nhttps://etherpad.opendev.org/p/inspur-fpga-driver-test-report.\nThis patch implemented Inspur FPGA driver in Cyborg. The spec is already\nmerged.\nPlease see:\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/victoria/approved/inspur-fpga-driver-proposal.html\n\nStory: 2007772\nTask: 39997\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a1ab5986d8bd264e97a4bbf19cfe9b76124339dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a1ab5986d8bd264e97a4bbf19cfe9b76124339dd"}]},"branch":"refs/heads/master"},"bccc1d95f4a0a1158eefe277e9c7580815151a3c":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-09-01 02:41:35.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/7"}}},"commit":{"parents":[{"commit":"40259a25ecc00520c19e88789b77306d8db1b272","subject":"Revert \"Add new default policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40259a25ecc00520c19e88789b77306d8db1b272"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-09-01 02:33:31.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nPlease see the test report in https://wiki.openstack.org/wiki/Cyborg/TestReport/InspurFPGA\nThis patch implemented Inspur FPGA driver in Cyborg. The spec is already\nmerged.\nPlease see:\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/victoria/approved/inspur-fpga-driver-proposal.html\n\nStory: 2007772\nTask: 39997\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bccc1d95f4a0a1158eefe277e9c7580815151a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bccc1d95f4a0a1158eefe277e9c7580815151a3c"}]},"branch":"refs/heads/master"},"ea18c5815be05c26ae0157501f7d17c26a42f0f1":{"kind":"REWORK","_number":8,"created":"2020-09-08 12:18:45.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/8"}}},"commit":{"parents":[{"commit":"40259a25ecc00520c19e88789b77306d8db1b272","subject":"Revert \"Add new default policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40259a25ecc00520c19e88789b77306d8db1b272"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-09-08 12:18:23.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nPlease see the test report in https://wiki.openstack.org/wiki/Cyborg/TestReport/InspurFPGA\nThis patch implemented Inspur FPGA driver in Cyborg. The spec is already\nmerged.\nPlease see:\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/victoria/approved/inspur-fpga-driver-proposal.html\n\nStory: 2007772\nTask: 39997\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ea18c5815be05c26ae0157501f7d17c26a42f0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ea18c5815be05c26ae0157501f7d17c26a42f0f1"}]},"branch":"refs/heads/master"},"3108d28eb3b50999d662765d5d8b90df760ff580":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-09-08 12:40:00.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/52/729852/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/52/729852/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/52/729852/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/52/729852/9"}}},"commit":{"parents":[{"commit":"40259a25ecc00520c19e88789b77306d8db1b272","subject":"Revert \"Add new default policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/40259a25ecc00520c19e88789b77306d8db1b272"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-21 07:04:57.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-09-08 12:37:06.000000000","tz":480},"subject":"Add Inspur FPGA driver","message":"Add Inspur FPGA driver\n\nPlease see the test report in https://wiki.openstack.org/wiki/Cyborg/TestReport/InspurFPGA\nThis patch implemented Inspur FPGA driver in Cyborg. Currently we don\u0027t\nsupport program with this card by Cyborg, the operator can bind the card\nto guest and use it in guest, so BSP data reported is ignored now.\n\nThe spec is already merged. Please see:\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/victoria/approved/inspur-fpga-driver-proposal.html\n\nStory: 2007772\nTask: 39997\n\nChange-Id: I7cc3482a4a6ac6fb0ca6965b7ce18bc37b122f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3108d28eb3b50999d662765d5d8b90df760ff580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3108d28eb3b50999d662765d5d8b90df760ff580"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
