)]}'
{"id":"openstack%2Fcyborg~749724","triplet_id":"openstack%2Fcyborg~master~I99f2f88c157f7e412b887ac8fc1abb19b735291a","project":"openstack/cyborg","branch":"master","topic":"policy-popup","hashtags":[],"change_id":"I99f2f88c157f7e412b887ac8fc1abb19b735291a","subject":"Add policies tests for existing device_profile APIs","status":"MERGED","created":"2020-09-03 12:36:58.000000000","updated":"2020-09-14 08:02:20.000000000","submitted":"2020-09-14 07:59:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749724-1600070383564-e73e0a8c","meta_rev_id":"b01e4bf405fa5ef9dceaf68632e7dfdfa48e1a0b","_number":749724,"virtual_id_number":749724,"owner":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 07:59:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"recommended":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2020-09-11 10:04:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2020-09-14 03:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2020-09-14 07:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2020-09-11 10:04:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-11 10:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":1,"date":"2020-09-14 07:25:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-05 10:39:10.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2020-09-07 11:48:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-09 01:10:12.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2020-09-11 04:00:01.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-09-11 05:52:41.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-09-11 08:41:37.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-09-14 03:47:54.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-09-14 07:25:23.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-09-14 07:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa52df1faa5c17c7ce2dbca60df07e88506edf08","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-03 12:36:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce7d0a286eb538c035e1c1f172bc0b2efd0e1d65","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-03 12:39:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"017adeb1813395a528621183de6e94824615c75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 13:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c17b385f37c45428ceb6ec43d306cef : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cdf34d9c684061842abf3d2ad9952d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c68818a608247548da9bdb704dfa862 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920f513c3e1a4236934cb72ca9595c7e : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e49c9e08d4641378e2e146de924fd6f : SUCCESS in 7m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ba71e3122f2a4a4b95e64262f1c7967f : SUCCESS in 31m 36s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3215115257594827991647b33294a503 : SUCCESS in 33m 58s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/ddfbb54d8c9941ee865a1ce91ac3c3a4 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"370c77f139f70456552af726742375184aa28ffb","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-04 01:51:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"cab36ce5196f6282cc781f492007af7d7634d254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 02:19:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cddc0367ee274ea0b03ef063d1d1d103 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedb65bfd42143dcbc86e3d6f15cafa9 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fff651e03e3d4ad792b4f0db46c65c6c : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691b850ff55d4fdda27d5449a8d7c9f9 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2910043ad4548dcbdaffaba8b064d8d : SUCCESS in 6m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d3a5933a3d8c4dc7894ad9bd99415219 : SUCCESS in 26m 02s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86bf3640fa7b4efc85467f5e9479aad6 : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3205595d28f27b0d1ef9baa199e3b2ef386abcf1","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-05 10:37:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5365b704c0f34820c3f1290001fffbeb9dd32110","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-05 10:39:35.000000000","message":"Topic set to policy-popup","accounts_in_message":[],"_revision_number":4},{"id":"049584ff25795c1782a5a7f27fbd8bede147a214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 10:58:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fa6c249175e47a8b19ae9ea0146acaf : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfeff9533c144883bfcb552766da279f : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/264a0f21b0af4ba9a4599a145bde7e75 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dce28a7d35ae4d7e88ede1f02da3d1e8 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4305dbe096ef4f159aa4c171a134f54d : SUCCESS in 5m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/987833de5d414d75a645b1a69863721e : SUCCESS in 19m 06s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db98fbd236f04893873ff8a0d1588f9a : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e32aaaae543c88e92335fb25d27923155dea12f8","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-07 07:01:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bdaacf273032dfa37b0ba560365b2074af332728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:37:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbc409898b6d4b63878433c6b8ffacff : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec9df78f0464884b44d8f546c00bd53 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d79427693e94c29bf1cffc9528bc92c : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/721831a3d1224814ae27737a11b6ab24 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8dd2186fcff45ed80f84edb009e9526 : SUCCESS in 8m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cafa4af0dbed48a08065979d93b312bf : SUCCESS in 27m 47s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec5f6e1d58a48068926587d67d75287 : SUCCESS in 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"52be9b5c45d8c0dbb96f9e2a1863bb6f7c82357d","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-08 09:53:22.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nLooks good to me overall. Soft -1 for one little question inline:)","accounts_in_message":[],"_revision_number":5},{"id":"128f877a90dbdbb8ae8d45aa233b45ac3bdb282c","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-09-09 01:10:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nAll LGTM. Please look at my comment in line.","accounts_in_message":[],"_revision_number":5},{"id":"f18f2d7c04b8a0525ce880610eb2932a964b7c91","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 02:56:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"6c6d464e7f852a73247379155450f0e8d88aed6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:31:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb019f9383c04916840f1c4fc60eae7e : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c66428de65894be3be18d1415fd0d695 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d527ffa97e3c4701ab3d8a5ef60669f3 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ede5cdcec60a463c9d97725fe55e018a : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37885f64a1f64ca8902a83c1d417b8c6 : SUCCESS in 17m 33s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ff9ee18da64e4a50a39f218b89750a0a : SUCCESS in 27m 51s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/effa13eac0564ba58aaf9f91ac5b0840 : SUCCESS in 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f95d32b006688f038a3301a9c77f1ec13680e90e","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 09:16:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3cab23ec944e9520a2bfc04d26f43b8da6c213b3","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 09:17:40.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"307494f2684ab99a931360184a177b3effbf3702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 09:55:58.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af71c20e86734455a384562ecb18a9e9 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a376768869de4a499ef4d074e48b97ce : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9261512dbc9d40d2803cd1ae2ac563ad : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0838c4495a8497b86da7aaaf10b7123 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70629cfe8b204efbae3af4681891c2e3 : SUCCESS in 18m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8e25548ba41740c9bef52600f4bc5c5e : SUCCESS in 30m 24s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34cef95a4cde46f0843fc80dac2e1438 : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eaa5e5de58c943bc9fbc2f4f4049d8701f6db2b1","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 12:11:36.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1dde4dca11939abbf0491fa217b317c14c4aaf48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 12:40:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79c59c0d2b854b008fa7af6dd079200a : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe80627fef1d49a2ba8a83cb783da5fc : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74b77e57223a4eafad0247ac5be1413c : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e62a0eb05e8a469397d61c8346cac66a : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99533aaac1b47499b6085c4b9ad8698 : SUCCESS in 14m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/005a606d5c6f4954bd060215a222f295 : SUCCESS in 26m 28s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1190be8408f44d9ab231674fe6c8ebfd : SUCCESS in 24m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"608f7a317acd7b705ac7304a4dad447c22d8c41c","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-11 00:42:29.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"06ba438882f39155d5ef003d2303ca1f463cefc9","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-09-11 04:00:01.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cce34dd1da12b96a36879ded9134346afdf61225","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-09-11 05:52:41.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":8},{"id":"8f4225d57f7319927abbf0c3ea6307f621a123fd","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-11 06:01:07.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1d96e771e562a4f2a73d3e7971cfacf7added04d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-11 08:41:37.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2274ac99699fc4df66e41d9ad7e5fabc2312e3bd","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-11 09:58:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7bf1f6e1ee3665dbd27db7cb24247fedefdae4b4","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-11 10:04:21.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"405b0c84cda86ebb4ab2be8b65a4ad03010300ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:43:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1febffd840834963a227f2ff4a3b9577 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7589e4d7d8764576b9f41ced70883231 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1df65f0d96647f892b7a5eab7c3f4a1 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201ca91a96624eab9a97ef23afdb2e7b : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50076ed1b52a42dca4c9484682623e2e : SUCCESS in 17m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/114e4463722b4028b74ae879174dda38 : SUCCESS in 28m 07s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8b48217535a44ab87080f6d060a42a7 : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c7056b084543f3a7c4e82f05b0eef184a5f8db74","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-14 03:47:54.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"3fe70c881a0f56d30c3d92205040627ec0a2e922","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-14 07:25:23.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f3c71fa6cd14b8f3a4d578ace1d4711be853cf15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 07:25:40.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"b6c3e4800a9346257783d33d977729edc7f85fd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 07:59:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"11bc383abf13a66645db556b37dfa3078192a16c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 07:59:43.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/742f04c3cc2c4d14b426e1b75819624e : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b3d72c9e8994c1582b3697453925403 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9dc51f73428409ba291595574b51402 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86c2995d2df44e6ae65a8755206bcf7 : SUCCESS in 32m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2923e0134fae4121b06a8ff090392c46 : SUCCESS in 22m 58s","accounts_in_message":[],"_revision_number":9},{"id":"b01e4bf405fa5ef9dceaf68632e7dfdfa48e1a0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 08:02:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5693a2f21ed4e39852b139458121114 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"27d5a7de3904e01a13aa450eba1045634a312d9c","revisions":{"4a4dcdbb12f0eb3cf53d5a3603f150c2a0206c90":{"kind":"REWORK","_number":1,"created":"2020-09-03 12:36:58.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/1"}}},"commit":{"parents":[{"commit":"767f2afdb4161d228247f19e591dfe948377368c","subject":"Define PolicyNotAuthorized exception and set 401 as unauthorized error_code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/767f2afdb4161d228247f19e591dfe948377368c"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"subject":"[WIP] Add tests for existing device_profile APIs","message":"[WIP] Add tests for existing device_profile APIs\n\nThis patch added tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4a4dcdbb12f0eb3cf53d5a3603f150c2a0206c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4a4dcdbb12f0eb3cf53d5a3603f150c2a0206c90"}]},"branch":"refs/heads/master"},"bc451c451dd70ab84b01bb6db460561b050d893d":{"kind":"REWORK","_number":2,"created":"2020-09-03 12:39:02.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/2"}}},"commit":{"parents":[{"commit":"767f2afdb4161d228247f19e591dfe948377368c","subject":"Define PolicyNotAuthorized exception and set 401 as unauthorized error_code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/767f2afdb4161d228247f19e591dfe948377368c"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:38:39.000000000","tz":480},"subject":"[WIP] Add tests for existing device_profile APIs","message":"[WIP] Add tests for existing device_profile APIs\n\nThis patch added tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bc451c451dd70ab84b01bb6db460561b050d893d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bc451c451dd70ab84b01bb6db460561b050d893d"}]},"branch":"refs/heads/master"},"6f4796b58aea1b29fb31d83e68daa23f4a0ed6f7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-04 01:51:50.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/3"}}},"commit":{"parents":[{"commit":"73871ba6693a7cc24a6303f3a49d35ab3281f05f","subject":"Define PolicyNotAuthorized exception and set 401 as unauthorized error_code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/73871ba6693a7cc24a6303f3a49d35ab3281f05f"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-04 01:51:50.000000000","tz":0},"subject":"[WIP] Add tests for existing device_profile APIs","message":"[WIP] Add tests for existing device_profile APIs\n\nThis patch added tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6f4796b58aea1b29fb31d83e68daa23f4a0ed6f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6f4796b58aea1b29fb31d83e68daa23f4a0ed6f7"}]},"branch":"refs/heads/master"},"ff71ae34ee418a64483d21ec9a59ba3e2594c04f":{"kind":"REWORK","_number":4,"created":"2020-09-05 10:37:18.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/4"}}},"commit":{"parents":[{"commit":"07aee5014d37b1683b4540e8c0afc785590956ae","subject":"Merge \"Add releasenote for add project_id for arq patch api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07aee5014d37b1683b4540e8c0afc785590956ae"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-05 10:36:39.000000000","tz":480},"subject":"Add policies tests for existing device_profile APIs","message":"Add policies tests for existing device_profile APIs\n\nThis patch added policies tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ff71ae34ee418a64483d21ec9a59ba3e2594c04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ff71ae34ee418a64483d21ec9a59ba3e2594c04f"}]},"branch":"refs/heads/master"},"407b96603ad01cb89c29302a8cbfc3eb308534e7":{"kind":"REWORK","_number":5,"created":"2020-09-07 07:01:16.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/5"}}},"commit":{"parents":[{"commit":"07aee5014d37b1683b4540e8c0afc785590956ae","subject":"Merge \"Add releasenote for add project_id for arq patch api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07aee5014d37b1683b4540e8c0afc785590956ae"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-07 07:00:56.000000000","tz":480},"subject":"Add policies tests for existing device_profile APIs","message":"Add policies tests for existing device_profile APIs\n\nThis patch added policies tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/407b96603ad01cb89c29302a8cbfc3eb308534e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/407b96603ad01cb89c29302a8cbfc3eb308534e7"}]},"branch":"refs/heads/master"},"0ad4e8c8bd843eb0bdb17d73903c1f36102d0797":{"kind":"NO_CHANGE","_number":6,"created":"2020-09-10 02:56:59.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/6"}}},"commit":{"parents":[{"commit":"07aee5014d37b1683b4540e8c0afc785590956ae","subject":"Merge \"Add releasenote for add project_id for arq patch api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07aee5014d37b1683b4540e8c0afc785590956ae"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 07:34:09.000000000","tz":480},"subject":"Add policies tests for existing device_profile APIs","message":"Add policies tests for existing device_profile APIs\n\nThis patch added policies tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0ad4e8c8bd843eb0bdb17d73903c1f36102d0797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0ad4e8c8bd843eb0bdb17d73903c1f36102d0797"}]},"branch":"refs/heads/master"},"1b703a81a450a1f46a1d0b2802da480829c112ad":{"kind":"REWORK","_number":7,"created":"2020-09-10 09:16:03.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/7"}}},"commit":{"parents":[{"commit":"21c43211d301cf4a9fd261af89ac6b92f457719c","subject":"Merge \"Add Inspur FPGA driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21c43211d301cf4a9fd261af89ac6b92f457719c"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-10 09:15:44.000000000","tz":480},"subject":"Add policies tests for existing device_profile APIs","message":"Add policies tests for existing device_profile APIs\n\nThis patch added policies tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1b703a81a450a1f46a1d0b2802da480829c112ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1b703a81a450a1f46a1d0b2802da480829c112ad"}]},"branch":"refs/heads/master"},"f42ec8d251dbcb1c718424c4119e6d7a5c9a39ed":{"kind":"REWORK","_number":8,"created":"2020-09-10 12:11:36.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/8"}}},"commit":{"parents":[{"commit":"21c43211d301cf4a9fd261af89ac6b92f457719c","subject":"Merge \"Add Inspur FPGA driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21c43211d301cf4a9fd261af89ac6b92f457719c"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-10 12:11:31.000000000","tz":0},"subject":"Add policies tests for existing device_profile APIs","message":"Add policies tests for existing device_profile APIs\n\nThis patch added policies tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f42ec8d251dbcb1c718424c4119e6d7a5c9a39ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f42ec8d251dbcb1c718424c4119e6d7a5c9a39ed"}]},"branch":"refs/heads/master"},"27d5a7de3904e01a13aa450eba1045634a312d9c":{"kind":"NO_CHANGE","_number":9,"created":"2020-09-11 10:04:21.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/24/749724/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/24/749724/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/24/749724/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/24/749724/9"}}},"commit":{"parents":[{"commit":"21c43211d301cf4a9fd261af89ac6b92f457719c","subject":"Merge \"Add Inspur FPGA driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21c43211d301cf4a9fd261af89ac6b92f457719c"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-03 12:34:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 06:26:58.000000000","tz":480},"subject":"Add policies tests for existing device_profile APIs","message":"Add policies tests for existing device_profile APIs\n\nThis patch added policies tests for existing device_profile APIs\n\nChange-Id: I99f2f88c157f7e412b887ac8fc1abb19b735291a\nStory: 2007024\nTask: 40313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/27d5a7de3904e01a13aa450eba1045634a312d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/27d5a7de3904e01a13aa450eba1045634a312d9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
