)]}'
{"id":"openstack%2Fcyborg~750863","triplet_id":"openstack%2Fcyborg~master~Ib58e6ba92513245dac915dfff29b02c556b542ee","project":"openstack/cyborg","branch":"master","topic":"policy-popup","hashtags":[],"change_id":"Ib58e6ba92513245dac915dfff29b02c556b542ee","subject":"Introduce scope_types in Device Profiles Actions","status":"MERGED","created":"2020-09-10 02:56:59.000000000","updated":"2020-09-16 07:00:50.000000000","submitted":"2020-09-16 06:58:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750863-1600239515960-72df3f54","meta_rev_id":"ac17179e19e17683d845a1ab1c8e1e36502819ad","_number":750863,"virtual_id_number":750863,"owner":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"date":"2020-09-16 05:52:08.000000000","_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 06:58:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"recommended":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2020-09-14 11:36:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2020-09-14 11:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2020-09-14 11:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":1,"date":"2020-09-14 11:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 09:42:14.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2020-09-10 09:42:19.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-09-10 09:42:31.000000000","updated_by":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"reviewer":{"_account_id":30759,"name":"Shogo Saito","email":"shogo.saito.ac@hco.ntt.co.jp","username":"s.shogo"},"state":"REVIEWER"},{"updated":"2020-09-11 03:59:32.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-09-11 08:40:31.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-09-14 03:46:44.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-09-14 08:52:38.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-09-16 06:58:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c660fb6d3e3f177d5b9fa6838562815c91a5dc4a","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 02:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"673916b157cbaac710528872c9699eacf8c8b38e","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 02:58:12.000000000","message":"Topic set to policy-popup","accounts_in_message":[],"_revision_number":1},{"id":"468991cb77e9852d57a45b44eaa557e66167a397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:22:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f15ab55bde4cc5bd9cb6daf7183191 : FAILURE in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b9faf04b214cfdbdcdd72052906a13 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0e29b62eacd46a2961d42cfa7482fcd : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75506680cc474098991aee7dc9e18423 : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914efb12a3e049d2b07edd4cf6f254ae : SUCCESS in 15m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/38378a5c7f75476b956386fa245eb0d6 : SUCCESS in 20m 51s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a894a48315d4847bd82e2b894f680b0 : SUCCESS in 22m 47s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/795a540d62814577ad01d3c0274d3869 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"27583c0f248a6aeb58fb842a4dc4df9785792785","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 08:22:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8515076e5ac206ad9516a290c19cdb55a794855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 09:07:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bfd91835f034eb9b90830def5725f3c : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60e733cdea14497a7dcce1289d41747 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13817090f3db4634832467e3120adc3b : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9414465d6e764ac0af3ffff4a9e52b79 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6709dc67ffe84330952f52b57c3a14f2 : SUCCESS in 15m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5b24d9d49e3c4ffd9e29b8843bae24de : SUCCESS in 31m 59s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8995616ae3d240e4a12cf0aab08224c3 : SUCCESS in 33m 59s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/06fc0e89fd2243918b2f8bc3b0b0e37f : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":2},{"id":"269891624d95968d8eb3669f5dbe8af5a9d56559","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 09:41:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d69e9948844d1c050d83f5b8a7464131f52f91d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 10:23:08.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc0cc54bd33242c18014abbc4095f522 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59135a61e323457ea5eddfa5168a64d6 : FAILURE in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/091f9e1557db42c4aa382b3f234c0ed8 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1358d63317bb41ce8b6c633efadef0a0 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fe2664d4244098a70a446cd707208b : SUCCESS in 14m 06s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2d6a27843dd34d039402642c4008d0b8 : SUCCESS in 31m 54s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54c434eb792843689ab3f205066f3cf4 : SUCCESS in 32m 23s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/5266a4daf2d04f1ebb022f69a288b40c : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"c0954f7cca485c9b0b5fb79f342cab07ba483e9b","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-10 12:15:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"10d1c4f9a638afe621714a210fe08e45f91e5178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 12:54:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94703f49b26444f7b2e06c8768610f51 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583a09d75e9241d681e1d9eac13fb27d : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24dbeaf129db4bb884fe6b2383f54b30 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f5b236208ad4f15a55edf6a3fd1f0fb : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1299c7b9156b47bf9a72082cc0f9c4a0 : SUCCESS in 15m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e0cbe75c85bd4deca7147325810c0b33 : SUCCESS in 29m 53s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dca5b4999aa479dba9a752afd05c7db : SUCCESS in 32m 25s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/e9e26a194f25496da2a8b6c64f982b31 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"e5fc73c82b2e29a5782ecbe6aba0f2905cf4a611","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-11 03:03:01.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3a10f129257996804e4d3eb38e1533a159d531ef","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-09-11 03:59:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f5de21e935f878270209f52f85c652530111f7a0","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-11 05:39:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1d2ada500a688df08ec2c187f76ffaea0925af1","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-11 08:40:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4c810dc3ad96963a76db0b92ea0e312722f5b5d1","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-11 10:04:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d39dad3178fd78a43ac1ea6fd53b01ce4f662bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:40:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65f1a8ab95174dfa974d97757d43dd46 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14969eaa2be4e87b6846b411e1a921f : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b23de841eb4348cc8fce17f224cdc674 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19c2dbbb22054603adb33ffa948fec9d : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e47b65e5a545729d28170d46f94959 : SUCCESS in 19m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/03f74eab94aa4b06ae3f278cf66ef9d3 : SUCCESS in 29m 37s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad7524bc65f14abb99fe6b674664f911 : SUCCESS in 28m 19s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/f01edb5d6be94009a6ed220b57864177 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":5},{"id":"4e4c7eb74e38765869ce9573c9760ceb5f6841fa","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-14 03:46:44.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"14dba0d95227d1dd2891e77a2fc90a173052f784","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 08:37:45.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3ff61420d7acfcba0a465cb416a69455278618e6","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-09-14 08:52:38.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"76cf8aa2f5d4aba467b56ffcfd644cb1df139bcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 09:38:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de3571d937ec4aee91a79879b68e5cc1 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ef1c0fdcfc4d59aee4f5c9486edade : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f1411aa8bae4a028c20a0ddc3d83419 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a4d570efaf64278a9f0bdab8c6b5c5b : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500acc63734c46f6a38b3970e191df55 : SUCCESS in 16m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1970ddf23ab04039b7ecaf2113fde8ca : SUCCESS in 46m 59s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e833da7e6f784c34bfd7a4fe40fba46d : SUCCESS in 49m 19s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/e35f5e8da2334973b358f0dd1e9973ed : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":6},{"id":"e3ac10c3eb8cafedc5ef19a574ead90bc2d2d6ce","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 11:36:35.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"4d2a72828d621ff3bd823970ac2154d015b598f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:14:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2fe49fccd014db68912fef49424ca5c : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3771284d8a346b497e3c9ec67fdf973 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e430e2a6f44b72be18d8828618bb34 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b24f72ba9145c7a7e33e0d57343307 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e517cc542d48a49cd67263f6fd929b : SUCCESS in 13m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0e91b21bf4424399a1be1fa42c010189 : SUCCESS in 30m 12s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5965a379dc24a399df66d5807e0ad09 : SUCCESS in 31m 47s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/7124fa7e054b4633a3aa7ecafed9bc7d : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":7},{"id":"76ddd470f1df1137e79bbfff91f1802d582b3b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 06:01:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cab168118056c7e53f13bf961bb42f56515ca6d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 06:40:41.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"3ced17587668c253ff4c9616642858229a5d137e","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-16 05:52:08.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"03c91cd3e8ab0a3302ac7adc040aa26d084fc278","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-09-16 06:03:13.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d6049adbe0e492318865020e15506fcc634cbcf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 06:28:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fcc038c8e6c4ff283a54e4db2c10f58 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6d6e4a144c4a73a3a0646a06b26fa8 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec0515a473af48c6982a60e22790f80e : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9819bd99d8249df946d7218d9109573 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c0b320b020428ea2f26135fb9c2ac5 : SUCCESS in 13m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8944dc5ca2bb4b0d90f87afdf0530f49 : SUCCESS in 30m 04s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37b4836811af4327ae9743b80ed7c0b7 : SUCCESS in 31m 54s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/fd5463f3738a418bb9034e5557bff6d0 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":7},{"id":"6154069aa404e0974da2cb910a1dd58004304c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 06:28:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a7bf5d48531e7bdac50aadcf000b88c1f53177d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 06:58:35.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d837f51e1457446b906c88d809ccb0bb : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7c9b745251e4eb3bce0a9b22fd32344 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/291d9ffc21e140858867d830b86f48c3 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f6bc72964c454fa013b936930f4348 : SUCCESS in 19m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5fe8eaf4099b410dbb250f5885ff62ea : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":7},{"id":"17668e052432c5b1a1fe367811d5c25e61394084","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 06:58:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"ac17179e19e17683d845a1ab1c8e1e36502819ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 07:00:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4acc52ec7f474db94039a81b8403d4 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9c1714b3ae837641aa3715b41d62bfe17edc7471","revisions":{"f502129870ccaa323794246bf79ac160baac18c9":{"kind":"REWORK","_number":1,"created":"2020-09-10 02:56:59.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/1"}}},"commit":{"parents":[{"commit":"349b7c4f19be6c9947512b2d1bc6db87d37d7199","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/349b7c4f19be6c9947512b2d1bc6db87d37d7199"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-09 12:19:36.000000000","tz":480},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f502129870ccaa323794246bf79ac160baac18c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f502129870ccaa323794246bf79ac160baac18c9"}]},"branch":"refs/heads/master"},"4569507c9905a001d070f3437dfb8cb8a0ddb3f4":{"kind":"REWORK","_number":2,"created":"2020-09-10 08:22:26.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/2"}}},"commit":{"parents":[{"commit":"349b7c4f19be6c9947512b2d1bc6db87d37d7199","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/349b7c4f19be6c9947512b2d1bc6db87d37d7199"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-10 08:20:49.000000000","tz":480},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4569507c9905a001d070f3437dfb8cb8a0ddb3f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4569507c9905a001d070f3437dfb8cb8a0ddb3f4"}]},"branch":"refs/heads/master"},"0175a86756b4a3e6d9771c6f7c669a6471f088fc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-10 09:41:40.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/3"}}},"commit":{"parents":[{"commit":"23498d237b1c9e0b41907e6b02d9927e79581932","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/23498d237b1c9e0b41907e6b02d9927e79581932"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-10 09:41:40.000000000","tz":0},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0175a86756b4a3e6d9771c6f7c669a6471f088fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0175a86756b4a3e6d9771c6f7c669a6471f088fc"}]},"branch":"refs/heads/master"},"7f2c79c9f76bd1d774fd5409bb2c4ad80d97ce11":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-10 12:15:41.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/4"}}},"commit":{"parents":[{"commit":"4a92acc916603858e909bb61df12d9750be18dff","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4a92acc916603858e909bb61df12d9750be18dff"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-10 12:15:41.000000000","tz":0},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7f2c79c9f76bd1d774fd5409bb2c4ad80d97ce11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7f2c79c9f76bd1d774fd5409bb2c4ad80d97ce11"}]},"branch":"refs/heads/master"},"5a4f668f0ff36a822347f46b1889ad8c74b5a204":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-11 10:04:21.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/5"}}},"commit":{"parents":[{"commit":"5a4ff80fe946986dd7713bef3479416677f91477","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5a4ff80fe946986dd7713bef3479416677f91477"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 06:27:50.000000000","tz":480},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5a4f668f0ff36a822347f46b1889ad8c74b5a204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5a4f668f0ff36a822347f46b1889ad8c74b5a204"}]},"branch":"refs/heads/master"},"91daa7102830729c50d8cffc302f568811af3257":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-14 08:37:45.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/6"}}},"commit":{"parents":[{"commit":"b1c223afd8fbd5821ab3dea26a33b28212165288","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b1c223afd8fbd5821ab3dea26a33b28212165288"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-14 08:37:45.000000000","tz":0},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/91daa7102830729c50d8cffc302f568811af3257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/91daa7102830729c50d8cffc302f568811af3257"}]},"branch":"refs/heads/master"},"9c1714b3ae837641aa3715b41d62bfe17edc7471":{"kind":"NO_CHANGE","_number":7,"created":"2020-09-14 11:36:35.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/63/750863/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/750863/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/750863/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/750863/7"}}},"commit":{"parents":[{"commit":"b1c223afd8fbd5821ab3dea26a33b28212165288","subject":"Add new default roles in Device Profile API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b1c223afd8fbd5821ab3dea26a33b28212165288"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-08 12:08:17.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-14 08:56:03.000000000","tz":480},"subject":"Introduce scope_types in Device Profiles Actions","message":"Introduce scope_types in Device Profiles Actions\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nEach policy rules will be covered with appropriate oslo.policy’s\n“scope_types”, ‘system’ and ‘project’ in cyborg cases as that defined in\nthe policies: https://wiki.openstack.org/wiki/Cyborg/Policy\n\nThis commit introduce scope_type for Device Profiles API policies.\n\nCreate and delete Device Profiles policies are scopped as \u0027system\u0027 because\ndevice_profile operation should not be given access to project scopped token,\nit has the same security requirement as that of manage a nova flavor.\n\nGET operations are scopped as [‘system’, ‘project’] because any\nreader(either system_reader or project_reader) can retrieve a device profile.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nStory: 2007024\nTask: 40836\nChange-Id: Ib58e6ba92513245dac915dfff29b02c556b542ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9c1714b3ae837641aa3715b41d62bfe17edc7471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9c1714b3ae837641aa3715b41d62bfe17edc7471"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
