)]}'
{"id":"openstack%2Fcyborg~751227","triplet_id":"openstack%2Fcyborg~master~I08f9d9043a488c8bb55c57f14f8eac860d47eb1f","project":"openstack/cyborg","branch":"master","topic":"policy-popup","hashtags":[],"change_id":"I08f9d9043a488c8bb55c57f14f8eac860d47eb1f","subject":"Add releasenote for policy refresh base and device profile policies","status":"MERGED","created":"2020-09-11 10:04:21.000000000","updated":"2020-10-10 07:26:01.000000000","submitted":"2020-10-10 07:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751227-1602314663150-2c0f4678","meta_rev_id":"0eb2eaa0b4df56b22f8ad902d55cf63bcf2e19f5","_number":751227,"virtual_id_number":751227,"owner":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-10 07:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2020-10-10 07:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2020-10-10 07:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":1,"date":"2020-10-10 07:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 10:06:07.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-09-16 18:51:30.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2020-09-25 02:48:26.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-09-29 15:24:10.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-10-01 19:01:17.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-10-10 07:02:25.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-10-10 07:04:50.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-10-10 07:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35f882ab72c8ad33b8618c3f744a82d95df97002","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-11 10:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b8511dfb268c0bad3da5dfc73e47bd57bbb56c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:30:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3551821adf5d46ee8ca256662f05c64b : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c163363d43b04fa09fac6aa7344ba857 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ac24ffcd1d4e0ab35d0a3c56203816 : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dfc86a53651a7955dfe044ba359e65e3ddd53b02","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 08:55:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e0199d2f1d9566d226858313d191ff41cfcb7d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 09:18:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1277e1e159444974aa8769c5f72f9dd5 : FAILURE in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fffae96f65ea4c0a9d13b6225cef8a3e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b1b3ef2d4a4071865721562ac4f08a : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fb9ac2a7965a1f4794fe4d1c77a8e346ff2c9bc8","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 11:36:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bfaa178b2567a68a87eee46c85fdf83c904b42a1","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 11:44:09.000000000","message":"Topic set to policy-popup","accounts_in_message":[],"_revision_number":3},{"id":"061e6092bc9fe78a22152a801f8c2261e8d87f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:59:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59c4e758a2a6453190311b378df6adc7 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b433b32e3d415f8e2c02bf0e69a601 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41dd5aa8ac94c84a89eb19c80b2b0d3 : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":3},{"id":"bd4db74d9dc01ec8936fea0580f40767ece23af4","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 12:22:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"615146c6a3bd2b6b318a037890bedde13ca1fe85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:42:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/636ef62660be461a9618ae10aa407b5a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fde45adb356b4a7ab838189e0c5d5865 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44dbe42d27b4f3b947685da95e603be : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":4},{"id":"5fe11dfd54b35067d1a2d5924b920e686c63245e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-15 02:44:53.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dce1fbb775a462dfd8de7b7f0d72c9ee4bd721ef","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-15 02:52:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ad214ad711b18c188192093c5abe05cb975ae77c","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-15 02:56:26.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nThanks Brin!\nPls see my update.","accounts_in_message":[],"_revision_number":5},{"id":"3a4d365d84a04de01773057c7112b5a7e0bd0b36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 03:06:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9b096aec5c84175b7fe259f05c82340 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57691cbd5bb1458e8748b69d7bca8ebf : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505d136ace24432badeb5fe74c95a6ee : SUCCESS in 12m 09s","accounts_in_message":[],"_revision_number":5},{"id":"e8367dc8f7a3ea19beb2b94fde4547bea895e0af","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-15 03:09:28.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"17aff8d5b837198fc4a2dfbf6f0f8aec6aad0e83","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-09-16 18:51:30.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ab9742a422e49ad74c7f990075a3d92f5f0784e4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-17 08:03:51.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nRaildo pointed was pointed by wenping in PS4","accounts_in_message":[],"_revision_number":5},{"id":"0c6c02afce6373e19a6b5ba499fcf1070a22297a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-17 21:35:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nnice work. let\u0027s finish the prework also before this.","accounts_in_message":[],"_revision_number":5},{"id":"e70ac8be97db7eda4a4e5722bd0efc1f58e1dcf9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-17 22:23:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5f5b0cd91bbae82e46841a3e9b275217e5d4ed0","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-24 03:54:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a2be126fc80cae4a1c9883940847c081711b1751","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-24 03:59:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"308ace813354c83b40687daba69604252a00b2af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:06:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2fdbf2ccb204f589b82a38eef2b0746 : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5019eb188af2432ab1d8527c8faff070 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff7c93ac7d743c38a8b52ade5db8e12 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":7},{"id":"5dd88034be5e892fa4f4bb3b0121e48861bb37ba","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-25 02:48:26.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"29a9cdf457b6a4af143488e7556fafaace6015fd","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-25 06:11:31.000000000","message":"Patch Set 7: Code-Review-1\n\n(10 comments)\n\nLooks good, but there are also some issues need to be fixed.","accounts_in_message":[],"_revision_number":7},{"id":"f727be49bb1e8bcc99e7818e14c3c80035f793c2","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-29 15:23:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e9f9c783fd6bb5588baaf539372c0567822d82e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 15:42:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7a58bb21eeb41228fc5ed52bc214376 : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6628b2fe08494169bb55a74a19a9644f : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a029080d73340718690f2b8c8d0fda8 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":8},{"id":"5a46df03dbbf836dc560823c5e8ab221148ce18a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-01 19:01:17.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nthanks for this, let\u0027s add the old deprecation rule in deprecation section.","accounts_in_message":[],"_revision_number":8},{"id":"6cf028a5810dc561b75966aa6a13dc593b334d59","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-10 06:46:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f29977c8894cac7be14f8c1ad037ea733a14a01f","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-10 06:47:11.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 0f323159c514ab2fd8b581da5bba4bcee7595481","accounts_in_message":[],"_revision_number":9},{"id":"da3d2701148b27bdf59ea0f94a8dab1e8cb8b4a3","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-10 06:48:11.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"912cd650171c0a62310bd196c6a4d233051f9551","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-10 06:55:19.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"7ef4694f40bebe2fa1712e1456ee48a189dafce4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-10 07:02:25.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"e90ace45679f9ff405c16ae1969c48dfa9b0403c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-10-10 07:04:50.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"8c2db3f6d72ec48e3d39061bb43f89b1fea17340","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-10-10 07:04:55.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"cddc033363caddfb7f2700c191cc8332535c68cf","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-10 07:07:20.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bae8f738e8498a32fecb552c103d3f57646dd84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 07:12:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bf95f1db7704a67a66b1980ad893850 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd5ef14a3b34dd896f25303ab153973 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b486250775f74b86be47e1efdaa7b4db : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":11},{"id":"ddf9518e89397b7e663c06a7db621ed3676a621a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 07:12:39.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"86d0fd662f1b09810b47e8c3f17240fc11b72e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 07:24:22.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80eed5e1bad4d45a04a389d8c7764b8 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f607256217eb439f95580796de28244d : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fad9f2e542e4cc2bf3a784dbcb2a95e : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":11},{"id":"3f34b13d93c314c71e9f5d3f56dbab9b4fa9eeac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 07:24:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"0eb2eaa0b4df56b22f8ad902d55cf63bcf2e19f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 07:26:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19884b5c10fc4e449f5150a12ec1d612 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da13083369ca4185ac9db8557981e2d6 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"bdc1aab530e18b86f39a102578c02a99a3c3abe0","revisions":{"163760ce8f6f07c03f6c620d1ab663eaf5ff16b1":{"kind":"REWORK","_number":1,"created":"2020-09-11 10:04:21.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/1"}}},"commit":{"parents":[{"commit":"5a4f668f0ff36a822347f46b1889ad8c74b5a204","subject":"Introduce scope_types in Device Profiles Actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5a4f668f0ff36a822347f46b1889ad8c74b5a204"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/163760ce8f6f07c03f6c620d1ab663eaf5ff16b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/163760ce8f6f07c03f6c620d1ab663eaf5ff16b1"}]},"branch":"refs/heads/master"},"60e5588e2065c4f27956df13150b1883acbbd3b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-14 08:55:39.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/2"}}},"commit":{"parents":[{"commit":"91daa7102830729c50d8cffc302f568811af3257","subject":"Introduce scope_types in Device Profiles Actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/91daa7102830729c50d8cffc302f568811af3257"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-14 08:55:39.000000000","tz":0},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/60e5588e2065c4f27956df13150b1883acbbd3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/60e5588e2065c4f27956df13150b1883acbbd3b0"}]},"branch":"refs/heads/master"},"f9ccadb508357a5824acb63aaade7e1ed764754d":{"kind":"REWORK","_number":3,"created":"2020-09-14 11:36:35.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/3"}}},"commit":{"parents":[{"commit":"9c1714b3ae837641aa3715b41d62bfe17edc7471","subject":"Introduce scope_types in Device Profiles Actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9c1714b3ae837641aa3715b41d62bfe17edc7471"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-14 11:35:03.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f9ccadb508357a5824acb63aaade7e1ed764754d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f9ccadb508357a5824acb63aaade7e1ed764754d"}]},"branch":"refs/heads/master"},"286100014b82b5ad510e95e5585fc5eea9b66efd":{"kind":"REWORK","_number":4,"created":"2020-09-14 12:22:49.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/4"}}},"commit":{"parents":[{"commit":"9c1714b3ae837641aa3715b41d62bfe17edc7471","subject":"Introduce scope_types in Device Profiles Actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9c1714b3ae837641aa3715b41d62bfe17edc7471"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-14 12:22:17.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/286100014b82b5ad510e95e5585fc5eea9b66efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/286100014b82b5ad510e95e5585fc5eea9b66efd"}]},"branch":"refs/heads/master"},"d824540b087083e4fcae5c488f5806a11c0805ef":{"kind":"REWORK","_number":5,"created":"2020-09-15 02:52:55.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/5"}}},"commit":{"parents":[{"commit":"9c1714b3ae837641aa3715b41d62bfe17edc7471","subject":"Introduce scope_types in Device Profiles Actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9c1714b3ae837641aa3715b41d62bfe17edc7471"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-15 02:51:33.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d824540b087083e4fcae5c488f5806a11c0805ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d824540b087083e4fcae5c488f5806a11c0805ef"}]},"branch":"refs/heads/master"},"140b6d4f192bb7ebf26e0c993beb4aaac9bea1aa":{"kind":"REWORK","_number":6,"created":"2020-09-24 03:54:24.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/6"}}},"commit":{"parents":[{"commit":"8e82aa29415d854a094722b47792bb5b9529731a","subject":"add document for cyborg new policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8e82aa29415d854a094722b47792bb5b9529731a"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-24 03:53:41.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/140b6d4f192bb7ebf26e0c993beb4aaac9bea1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/140b6d4f192bb7ebf26e0c993beb4aaac9bea1aa"}]},"branch":"refs/heads/master"},"a6c5e7af023a7954d3b16f649a57f87e09df94ce":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-24 03:59:03.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/7"}}},"commit":{"parents":[{"commit":"5997bdcc1fc809dddb9638cd8947f906cabdf9f4","subject":"add document for cyborg new policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5997bdcc1fc809dddb9638cd8947f906cabdf9f4"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-09-24 03:59:03.000000000","tz":0},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a6c5e7af023a7954d3b16f649a57f87e09df94ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a6c5e7af023a7954d3b16f649a57f87e09df94ce"}]},"branch":"refs/heads/master"},"07ebfce8d40485ff1aab86cdc090933b5d65ac6b":{"kind":"REWORK","_number":8,"created":"2020-09-29 15:23:38.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/8"}}},"commit":{"parents":[{"commit":"665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3","subject":"Merge \"Add rc check for POST Device Profile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-29 15:23:04.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07ebfce8d40485ff1aab86cdc090933b5d65ac6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/07ebfce8d40485ff1aab86cdc090933b5d65ac6b"}]},"branch":"refs/heads/master"},"1c3e19f0f04bd330bc31321ea5ff7ea5a6cfbf89":{"kind":"REWORK","_number":9,"created":"2020-10-10 06:46:26.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/9"}}},"commit":{"parents":[{"commit":"66fea7ddcf096d2e628f017518b4046c1a770b3e","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/66fea7ddcf096d2e628f017518b4046c1a770b3e"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-10 06:46:05.000000000","tz":480},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1c3e19f0f04bd330bc31321ea5ff7ea5a6cfbf89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1c3e19f0f04bd330bc31321ea5ff7ea5a6cfbf89"}]},"branch":"refs/heads/master"},"ca3cb332f41804ed672e369a16d16894a089fada":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-10-10 06:48:11.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/27/751227/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/10"}}},"commit":{"parents":[{"commit":"c1f4e3806da06189566434129c5969c33ff1f9a4","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c1f4e3806da06189566434129c5969c33ff1f9a4"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-10-10 06:48:11.000000000","tz":0},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ca3cb332f41804ed672e369a16d16894a089fada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ca3cb332f41804ed672e369a16d16894a089fada"}]},"branch":"refs/heads/master"},"bdc1aab530e18b86f39a102578c02a99a3c3abe0":{"kind":"REWORK","_number":11,"created":"2020-10-10 06:55:19.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/27/751227/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/27/751227/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/27/751227/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/27/751227/11"}}},"commit":{"parents":[{"commit":"c1f4e3806da06189566434129c5969c33ff1f9a4","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c1f4e3806da06189566434129c5969c33ff1f9a4"}]}],"author":{"name":"Yumeng Bao","email":"yumeng_bao@yahoo.com","date":"2020-09-11 10:02:57.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-10-10 06:55:13.000000000","tz":0},"subject":"Add releasenote for policy refresh base and device profile policies","message":"Add releasenote for policy refresh base and device profile policies\n\nChange-Id: I08f9d9043a488c8bb55c57f14f8eac860d47eb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bdc1aab530e18b86f39a102578c02a99a3c3abe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bdc1aab530e18b86f39a102578c02a99a3c3abe0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
