)]}'
{"id":"openstack%2Fcyborg~754681","triplet_id":"openstack%2Fcyborg~master~Ic9f140f042d92baf34d326ad45505ede6da475af","project":"openstack/cyborg","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"Ic9f140f042d92baf34d326ad45505ede6da475af","subject":"Remove Six","status":"ABANDONED","created":"2020-09-28 10:13:46.000000000","updated":"2020-10-08 09:07:40.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7782806802723439cf390c38c9621453090aba89","_number":754681,"virtual_id_number":754681,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-30 07:44:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 07:44:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"761d2c2c5a0e25cf28f6fcb8b6747587dc6dcf05","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-28 10:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"478fdf0aee7052ee85d2d59fc041457d946aa8d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:53:18.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d77cfaf2f10540d1b1a72ed409b05aab : FAILURE in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1396b24fa66e4763bed64f0ddbcc4624 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f279d45e302549f9aa8cbb810de794f9 : FAILURE in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/963e35cc7435458d903090d1c301e45d : FAILURE in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d81bfb52b243404e91152a7ad6aa1d08 : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b68c3f820c49218bb4710639f58b14 : SUCCESS in 5m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc438d7060074ddcbc09a0dd5798b999 : FAILURE in 26m 01s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6acedd6e25414f758a314d5e52e59409 : FAILURE in 32m 20s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/9d518a6a3ff642a899eaf1262c737934 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6d2759adede4167a05e02a7ae88bdd7f04820412","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 06:51:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff9bf10896902e63cd8c0463750dc67bafb39ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 07:44:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e5fa525a924784a3a7945cdbe49132 : SUCCESS in 5m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aed4992a029549e59ac2e448549d4209 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7482c710d4464eb01a77b50ab19924 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ccb33e2550946b9b760aaa1c05a1a62 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4bf3f7bdb54b3d94f6ec7b7dfc2a2a : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9276a4a0bf694da48a8f67eb8f7955e5 : SUCCESS in 6m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2dfc24ba4db247d79ca80a97b62daaea : SUCCESS in 30m 42s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a923e609de5a48bd9f77f4d069837127 : SUCCESS in 37m 01s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/c8ff4582e404475e897787228d2a387d : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a4160cbab41d6321efba1fc191c8ca3aefd4de7d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 19:27:07.000000000","message":"Topic set to bp/six-removal","accounts_in_message":[],"_revision_number":2},{"id":"7782806802723439cf390c38c9621453090aba89","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-08 09:07:40.000000000","message":"Abandoned\n\nrepeat","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9cce203aff71bc06896b14566c94ff3307ad1065","revisions":{"f5a02821a8f539adf9f1686df1d72a3e456933f4":{"kind":"REWORK","_number":1,"created":"2020-09-28 10:13:46.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/81/754681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/754681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/754681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/754681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/754681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/754681/1"}}},"commit":{"parents":[{"commit":"665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3","subject":"Merge \"Add rc check for POST Device Profile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-28 10:13:28.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-28 10:13:28.000000000","tz":480},"subject":"Remove Six Replace the following items with Python 3 style code.","message":"Remove Six\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.PY2\n- six.PY3\n- six.text_type\n- six.string_types\n- six.ensure_str\n- six.integer_types\n\nChange-Id: Ic9f140f042d92baf34d326ad45505ede6da475af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f5a02821a8f539adf9f1686df1d72a3e456933f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f5a02821a8f539adf9f1686df1d72a3e456933f4"}]},"branch":"refs/heads/master"},"9cce203aff71bc06896b14566c94ff3307ad1065":{"kind":"REWORK","_number":2,"created":"2020-09-30 06:51:26.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/81/754681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/81/754681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/754681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/754681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/81/754681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/81/754681/2"}}},"commit":{"parents":[{"commit":"665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3","subject":"Merge \"Add rc check for POST Device Profile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-28 10:13:28.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 06:51:00.000000000","tz":480},"subject":"Remove Six","message":"Remove Six\n\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.PY2\n- six.PY3\n- six.text_type\n- six.string_types\n- six.ensure_str\n- six.integer_types\n\nChange-Id: Ic9f140f042d92baf34d326ad45505ede6da475af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9cce203aff71bc06896b14566c94ff3307ad1065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9cce203aff71bc06896b14566c94ff3307ad1065"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
