)]}'
{"id":"openstack%2Fcyborg~756976","triplet_id":"openstack%2Fcyborg~master~I0ef7139ded8f0218f037f1a532744b4d37f441bc","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"I0ef7139ded8f0218f037f1a532744b4d37f441bc","subject":"Add project_id description in api_ref","status":"MERGED","created":"2020-10-09 02:26:51.000000000","updated":"2020-11-19 07:16:45.000000000","submitted":"2020-11-19 07:14:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756976-1605770086248-6db76d6a","meta_rev_id":"cfa347b6180aaeab94d42f6293e404a220f57f8f","_number":756976,"virtual_id_number":756976,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 07:14:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2020-11-19 05:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":1,"date":"2020-11-19 05:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 02:24:14.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-11-10 02:29:27.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-11-19 03:48:52.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-11-19 05:55:17.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-11-19 07:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86276239f38d6059a11a406bc3885ff1cacfcd14","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-09 02:26:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f0b75190142a320f5903eac1a78be9cb7fc768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f37333a4964f4728bde7fe1f476590eb : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5594962e20d14260b725d03f85502cb8 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e49d15528264833a22370dd42e2734d : SUCCESS in 5m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f959bed75b784d569fdd7755103e8101 : SUCCESS in 28m 18s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1aca2665af042d68ca569652982e7b3 : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"548b75d6f6b273f411d36f8105a4d45a10c935a8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-13 00:44:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"67b6cc20f7293d1ab67e488b26b97b071391d2f6","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-04 09:35:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25cc2c2749666f91872b4e728d39e3954629fec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:28:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2ce09d8755bf4308a648f4a4c09cbee9 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55c03b7dec9545b985bcee23fd193625 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2eb5b1080c4c53997ad10df211b362 : SUCCESS in 5m 06s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c789d950bce64caaa95785c9936e8c7d : SUCCESS in 25m 27s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6054c987f10d46488f16a47d5bd68432 : SUCCESS in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e79f0b0bd8ac0daac0fce074b6917c2d99f9e506","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-09 05:40:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\n+1, please change the comment inline.","accounts_in_message":[],"_revision_number":2},{"id":"132aa1151ae55087701dc2ead6a422e19b32991b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-19 01:10:29.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3c6cea5b7482b9ca59e75eef46438ef3ab614c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 01:48:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/31a9fb666d1640fcac30ab7497973a59 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0384d6a97b2643849c62d171b6286dfc : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc45301d9c934753bba875b4a86bd01f : SUCCESS in 5m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/640660e236644dc8aea349d60d193e22 : SUCCESS in 35m 37s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f8b92597e4142cbb655a97dedfd246d : SUCCESS in 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aced616bba2506e3158fa11178b14abb2953d0c9","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-19 03:48:52.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"0da20cda034d37dfa74386d55cb3a44dfca10154","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-19 05:53:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"bcd334f58e4877f92ed0d9660f3eead76e85eb87","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-19 05:55:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)\n\nFixed my comments in PS2.\n\nnit change, +2 before by xinranwang, so +2 and +1W.\n\nThanks.","accounts_in_message":[],"_revision_number":4},{"id":"1a53738d6c9b77f77eb2189547c3166c0ef96bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 06:33:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d7edb6f07adf40b68e19467735491bf0 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a652eb84011b418b889108b764a5679b : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd0e4bf9d4e45b28be2e0fb2667ce7d : SUCCESS in 5m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ed61931133a146bca6d37462fe8bc58f : SUCCESS in 33m 43s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f91b4f38d68143fe971be4866a70b610 : SUCCESS in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d974000bec9ea9dfc489a9c0cb9e8a136b73aebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 06:34:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"640f7d1d5ee3d51bcb5f9fb93b67755c21563dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:14:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"06d112402b6318e7c3a2a57ea09162bf0f9e3e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:14:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59851d0186e5483ba813bab825773de8 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b81e8d9ea714792b842541988c63ed7 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8d35136b4d4bcc8a5ea5813724db1d : SUCCESS in 7m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/221a18f66416456cbe8366e028b4da1b : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":4},{"id":"cfa347b6180aaeab94d42f6293e404a220f57f8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:16:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cbb61a8f32af46e2bcab207194d09f5c : SUCCESS in 1m 25s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600eb7c2e25a45ef8cd3be12d917f0a0 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a069962852bdabc4b8c2d82c67554385b7d860f2","revisions":{"ece01daa0162cc145a9bdb853899fd7c782a8445":{"kind":"REWORK","_number":1,"created":"2020-10-09 02:26:51.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/76/756976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/76/756976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/76/756976/1"}}},"commit":{"parents":[{"commit":"665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3","subject":"Merge \"Add rc check for POST Device Profile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/665abe1f00b0a39fac5b9fdc4ef1f749c3d62db3"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-09 02:26:15.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-09 02:26:21.000000000","tz":480},"subject":"Add project_id description in api_ref","message":"Add project_id description in api_ref\n\nChange-Id: I0ef7139ded8f0218f037f1a532744b4d37f441bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ece01daa0162cc145a9bdb853899fd7c782a8445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ece01daa0162cc145a9bdb853899fd7c782a8445"}]},"branch":"refs/heads/master"},"76df561fa42cf9b037510ac24654a1a9e060533d":{"kind":"REWORK","_number":2,"created":"2020-11-04 09:35:26.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/76/756976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/76/756976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/76/756976/2"}}},"commit":{"parents":[{"commit":"02180ece20fe4982d084e1724d86171c2cb71e86","subject":"Merge \"Add fast8 check for cybrog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/02180ece20fe4982d084e1724d86171c2cb71e86"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-09 02:26:15.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-11-04 09:32:49.000000000","tz":480},"subject":"Add project_id description in api_ref","message":"Add project_id description in api_ref\n\nProject_id is added for arq, so update the api_ref correspondingly.\n\nChange-Id: I0ef7139ded8f0218f037f1a532744b4d37f441bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/76df561fa42cf9b037510ac24654a1a9e060533d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/76df561fa42cf9b037510ac24654a1a9e060533d"}]},"branch":"refs/heads/master"},"06f205f5afa5e689810bea76602c79324a91d8c6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-19 01:10:29.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/76/756976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/76/756976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/76/756976/3"}}},"commit":{"parents":[{"commit":"02180ece20fe4982d084e1724d86171c2cb71e86","subject":"Merge \"Add fast8 check for cybrog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/02180ece20fe4982d084e1724d86171c2cb71e86"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-09 02:26:15.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-11-19 01:09:25.000000000","tz":480},"subject":"Add project_id description in api_ref","message":"Add project_id description in api_ref\n\nProject id is added for arq, so update the api_ref correspondingly.\n\ntask: 41290\nstory: 2008375\n\nChange-Id: I0ef7139ded8f0218f037f1a532744b4d37f441bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06f205f5afa5e689810bea76602c79324a91d8c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06f205f5afa5e689810bea76602c79324a91d8c6"}]},"branch":"refs/heads/master"},"a069962852bdabc4b8c2d82c67554385b7d860f2":{"kind":"REWORK","_number":4,"created":"2020-11-19 05:53:51.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/76/756976/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/76/756976/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/76/756976/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/76/756976/4"}}},"commit":{"parents":[{"commit":"02180ece20fe4982d084e1724d86171c2cb71e86","subject":"Merge \"Add fast8 check for cybrog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/02180ece20fe4982d084e1724d86171c2cb71e86"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-09 02:26:15.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-11-19 05:53:44.000000000","tz":0},"subject":"Add project_id description in api_ref","message":"Add project_id description in api_ref\n\nProject id is added for arq, so update the api_ref correspondingly.\n\ntask: 41290\nstory: 2008375\n\nChange-Id: I0ef7139ded8f0218f037f1a532744b4d37f441bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a069962852bdabc4b8c2d82c67554385b7d860f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a069962852bdabc4b8c2d82c67554385b7d860f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
