)]}'
{"id":"openstack%2Fcyborg~757512","triplet_id":"openstack%2Fcyborg~master~I0d2b344939426260f9ab8d3c0c9c316b0ee5380d","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"I0d2b344939426260f9ab8d3c0c9c316b0ee5380d","subject":"Use TOX_CONSTRAINTS_FILE instead of obsolete UPPER_CONSTRAINTS_FILE","status":"ABANDONED","created":"2020-10-12 09:24:36.000000000","updated":"2020-10-12 11:04:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"614a30d6afa98d6f4791ab1b5d23edeab689b572","_number":757512,"virtual_id_number":757512,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-12 09:43:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 09:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6516d9b4e52ab9da23e43f5aac2bf176396b011b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-12 09:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"450492f63de39518c5b7ca07ac77ba4e263736d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:43:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/401132ca31704372b7ee315b6ffeb8ee : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f713841bad64fc58312417c51d89d02 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee4c4aad78364528b4cd9b48e1275bfc : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ab35f09ee649748e4f3f6dbf58fd6c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9281341826854c16a44e18106dd80dae : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/572e75a542ac4f6bb2e8c9816a2d57fa : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1adf97f0acce4ed8aa1c5ac9d8d226ba : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"614a30d6afa98d6f4791ab1b5d23edeab689b572","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-12 11:04:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98dfd40f18e1ad43364c04a41c0c36d92113024b","revisions":{"98dfd40f18e1ad43364c04a41c0c36d92113024b":{"kind":"REWORK","_number":1,"created":"2020-10-12 09:24:36.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/12/757512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/12/757512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/757512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/757512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/12/757512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/12/757512/1"}}},"commit":{"parents":[{"commit":"aa8633a57eabae67222fac7c9ea08efcb704103e","subject":"Merge \"Bump py37 to py38 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/aa8633a57eabae67222fac7c9ea08efcb704103e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-12 09:24:22.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-12 09:24:22.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE instead of obsolete UPPER_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE instead of obsolete UPPER_CONSTRAINTS_FILE\n\nChange-Id: I0d2b344939426260f9ab8d3c0c9c316b0ee5380d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/98dfd40f18e1ad43364c04a41c0c36d92113024b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/98dfd40f18e1ad43364c04a41c0c36d92113024b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
