)]}'
{"id":"openstack%2Fcyborg~789491","triplet_id":"openstack%2Fcyborg~master~I197c8ea49b3f89739e569b589b40888b7d036f4d","project":"openstack/cyborg","branch":"master","topic":"bug/1927054","hashtags":[],"change_id":"I197c8ea49b3f89739e569b589b40888b7d036f4d","subject":"docs: Add parameters list to device profile APIs","status":"MERGED","created":"2021-05-04 07:07:15.000000000","updated":"2021-05-28 09:31:36.000000000","submitted":"2021-05-28 09:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"789491-bug/1927054","meta_rev_id":"2e1218b64b820e8a91f4466844e42b0c5a1b0dc7","_number":789491,"virtual_id_number":789491,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 09:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2021-05-28 09:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2021-05-14 13:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2021-05-28 09:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 07:41:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-13 02:39:13.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2021-05-28 07:31:55.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"ccd9454cc32fbbba15640b10371cd434f6fc39c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-04 07:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa6a1bfe4d4f2d8304531221654145f05d56b302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-04 07:11:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7a30b5e1f83b34729dc46703e558d3a73a747e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 07:41:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbdb48d0f81043deb2477f5e26703458 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3afac52a164ee7bf67a4d424757eb7 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a63de5da394d7f9fb7d945faa399f2 : SUCCESS in 6m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/25f85e5778f24a17b3d0f7286e8a22af : SUCCESS in 25m 18s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14bd48959f5a468daa21967f0705efa6 : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e59c3798f4bdd4a1b6cf15ae42fb4165ea752704","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-05-13 03:00:24.000000000","message":"Patch Set 2: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"b4499896f642272eeea5aebc3e5fbdc758d64b99","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-05-13 03:01:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aba4a8b9d4c1dc92331dcbeba6ec39eecca73689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-13 10:29:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a4103ab2ff773b61e708280c74f00059651fec7a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-13 10:30:18.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks, addressed.","accounts_in_message":[],"_revision_number":3},{"id":"f751cbeb940dbb0f36d23304d21863004814f642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:56:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/355c554a65964ab39112dc358ae50bb1 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9314415cbfbb48a9af790c02a71bec12 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade1d2c0b10041c599edef12f89416a2 : SUCCESS in 6m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/06a34348b0af4a7684aacb28e4cc997b : SUCCESS in 23m 34s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1a25527a1534fb09e1e791bac5f54d5 : SUCCESS in 25m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7433e2a5412f78e85dfb3aaf8a4c2af2a71f3c2e","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-05-14 13:57:47.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"11d431006bf18dea70e2c946124d03aefac9973d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-28 08:24:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23e7b97e1df23d9af24a2d760ce94b5622268106","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-28 08:44:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"551aa9cee11d2eb0e091029c6929a91d94d57eab","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-28 09:05:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"67297872b3dac5a1083f304adab6f141080bc917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:05:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"03afd215f4ebdda339b6af84101360c0b7b39553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:30:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64b40a3070c14984af4b8c6e229f3f9b : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b485187111540a094135d59df0aa047 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6c6474bf20408d84a1e25132bc1ed1 : SUCCESS in 7m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3d76cf61a6334b519ca462ef03a23465 : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":3},{"id":"1ce6c5dbc85cae4ab9b8990e043e50ce8d5f6001","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:30:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2e1218b64b820e8a91f4466844e42b0c5a1b0dc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:31:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b124633db93c4bd5bc1d07c20b4f6c84 : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7be4e923cb74700a5a30bce68b4fbd4 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7605982401253af416d7636c390d99008eddfad7","revisions":{"1d9c10c9beae915c95dcfb2e6d2c97378e3743cd":{"kind":"REWORK","_number":1,"created":"2021-05-04 07:07:15.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/91/789491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/91/789491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/91/789491/1"}}},"commit":{"parents":[{"commit":"5bd8d929a719f2263c12c8ed5a4d51d2d40b6d3d","subject":"Merge \"Fix wrong args when calling unbind() method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5bd8d929a719f2263c12c8ed5a4d51d2d40b6d3d"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-05-04 07:02:56.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-05-04 07:02:59.000000000","tz":480},"subject":"docs: Add parameters list to device profile APIs","message":"docs: Add parameters list to device profile APIs\n\nSince we were introduced microversion in Ussuri [1], that if we need to\nadd a microversion for an API we should mark it as\nmin_version and/or max_version, so we should maintain these parameters\nwith a list in the docs, it will be clearer and friendly for users.\n\n[1]https://specs.openstack.org/openstack/cyborg-specs/specs/ussuri/implemented/api-microversion.html\n\nChange-Id: I197c8ea49b3f89739e569b589b40888b7d036f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1d9c10c9beae915c95dcfb2e6d2c97378e3743cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1d9c10c9beae915c95dcfb2e6d2c97378e3743cd"}]},"branch":"refs/heads/master"},"343a9bd9eb5a427c34f98a5cc475aad04298c939":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-04 07:11:06.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/91/789491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/91/789491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/91/789491/2"}}},"commit":{"parents":[{"commit":"5bd8d929a719f2263c12c8ed5a4d51d2d40b6d3d","subject":"Merge \"Fix wrong args when calling unbind() method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5bd8d929a719f2263c12c8ed5a4d51d2d40b6d3d"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-05-04 07:02:56.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-05-04 07:10:33.000000000","tz":480},"subject":"docs: Add parameters list to device profile APIs","message":"docs: Add parameters list to device profile APIs\n\nSince we were introduced microversion in Ussuri [1], that if we need to\nadd a microversion for an API we should mark it as\nmin_version and/or max_version, so we should maintain these parameters\nwith a list in the docs, it will be clearer and friendly for users.\n\n[1]https://specs.openstack.org/openstack/cyborg-specs/specs/ussuri/implemented/api-microversion.html\n\nCloses-Bugs: #1927054\n\nChange-Id: I197c8ea49b3f89739e569b589b40888b7d036f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/343a9bd9eb5a427c34f98a5cc475aad04298c939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/343a9bd9eb5a427c34f98a5cc475aad04298c939"}]},"branch":"refs/heads/master"},"7605982401253af416d7636c390d99008eddfad7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-05-13 10:29:46.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/91/789491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/91/789491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/91/789491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/91/789491/3"}}},"commit":{"parents":[{"commit":"5bd8d929a719f2263c12c8ed5a4d51d2d40b6d3d","subject":"Merge \"Fix wrong args when calling unbind() method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5bd8d929a719f2263c12c8ed5a4d51d2d40b6d3d"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-05-04 07:02:56.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2021-05-13 10:29:39.000000000","tz":0},"subject":"docs: Add parameters list to device profile APIs","message":"docs: Add parameters list to device profile APIs\n\nSince we were introduced microversion in Ussuri [1], that if we need to\nadd a microversion for an API we should mark it as\nmin_version and/or max_version, so we should maintain these parameters\nwith a list in the docs, it will be clearer and friendly for users.\n\n[1]https://specs.openstack.org/openstack/cyborg-specs/specs/ussuri/implemented/api-microversion.html\n\nCloses-Bug: #1927054\n\nChange-Id: I197c8ea49b3f89739e569b589b40888b7d036f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7605982401253af416d7636c390d99008eddfad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7605982401253af416d7636c390d99008eddfad7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
