)]}'
{"id":"openstack%2Fcyborg~804506","triplet_id":"openstack%2Fcyborg~stable%2Fvictoria~Ic21f9bc64038073a2e97dc906e09ef4bab042047","project":"openstack/cyborg","branch":"stable/victoria","hashtags":[],"change_id":"Ic21f9bc64038073a2e97dc906e09ef4bab042047","subject":"Resolve mysql conflict message","status":"ABANDONED","created":"2021-08-13 09:14:22.000000000","updated":"2021-08-17 06:16:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8522771682322d005c15566fcb02509f5473928d","_number":804506,"virtual_id_number":804506,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-13 09:47:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":-1,"date":"2021-08-15 06:39:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-13 09:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-13 14:47:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-08-15 06:39:19.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-08-16 00:41:51.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2021-08-16 00:41:51.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"}],"messages":[{"id":"ff6522f1fa3f1c0310b1e76b53efcdf91908eff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-08-13 09:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"372c2cd21b0f94e3283729953bf035a0d727ab07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 09:47:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8609c459154882b962a16952fb4a9d : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567f77e171024fff8b82767a7e47781b : FAILURE in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c276ba58d247cda3347b7fe648210a : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ae164ff76b945e5b2df5b56a6c2f068 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600523b01ddc4cab89a27bf8d4fc4132 : SUCCESS in 8m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8912772cffa243f8b9f37d804a20bc5a : SUCCESS in 31m 07s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17993616fae940f7b88c1269270a303e : SUCCESS in 26m 08s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/325d2bddf4dc4da0939e4f5429d7dae9 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d6d8fa588837db9c0c909e91fc2b084b6f35d8f0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-08-13 14:47:28.000000000","message":"Patch Set 1:\n\nthis is fixing the failure and we do not need to bump the version for stable.\n\ncan this be done in master (hoe this work for oslo.db(mysql) new version too) and then backport for all stable where they are failing?","accounts_in_message":[],"_revision_number":1},{"id":"c257904c345764072a31d1153171a9c4ac8f111a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-08-15 06:39:19.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e this is fixing the failure and we do not need to bump the version for stable.\n\u003e \n\u003e can this be done in master (hoe this work for oslo.db(mysql) new version too) and then backport for all stable where they are failing?\n\nYes, and https://review.opendev.org/c/openstack/cyborg/+/797403 this patch has been done this, we can backport after it merged.","accounts_in_message":[],"_revision_number":1},{"id":"cb0cb5f10d758b832d7891ec52452d06b96d1977","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-08-15 06:43:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e this is fixing the failure and we do not need to bump the version for stable.\n\u003e \u003e \n\u003e \u003e can this be done in master (hoe this work for oslo.db(mysql) new version too) and then backport for all stable where they are failing?\n\u003e \n\u003e Yes, and https://review.opendev.org/c/openstack/cyborg/+/797403 this patch has been done this, we can backport after it merged.\n\nSorry, above paste link is wrong. There is need to have a new patch and then backport it to stable/branch.\nI will submit a bug in launchpad to trace this issue.","accounts_in_message":[],"_revision_number":1},{"id":"eb993c606a041392ecbd2a6d947faf8b11fef5c8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-08-15 07:04:10.000000000","message":"Patch Set 1:\n\nhttps://bugs.launchpad.net/openstack-cyborg/+bug/1939970","accounts_in_message":[],"_revision_number":1},{"id":"d028052f8c4928b300498ccebdadc4fc7d24f215","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-08-16 00:49:05.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e this is fixing the failure and we do not need to bump the version for stable.\n\u003e \n\u003e can this be done in master (hoe this work for oslo.db(mysql) new version too) and then backport for all stable where they are failing?\n\nyes, gmman, we donnot need to bump the version for stable, but backport the code to stable. Devstack doesnot rely on the cyborg\u0027s package version.","accounts_in_message":[],"_revision_number":1},{"id":"8522771682322d005c15566fcb02509f5473928d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-08-17 06:16:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19559a4f552ee14a263ae5ca9e7d04e2947efee4","revisions":{"19559a4f552ee14a263ae5ca9e7d04e2947efee4":{"kind":"REWORK","_number":1,"created":"2021-08-13 09:14:22.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/06/804506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/06/804506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/804506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/804506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/06/804506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/06/804506/1"}}},"commit":{"parents":[{"commit":"32afabdfcbb7bcf348980b079849e0d985a2762c","subject":"Merge \"tools: Allow check-cherry-picks.sh to be disabled by an env var\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/32afabdfcbb7bcf348980b079849e0d985a2762c"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-08-13 09:13:48.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-08-13 09:13:48.000000000","tz":480},"subject":"Resolve mysql conflict message","message":"Resolve mysql conflict message\n\nChange-Id: Ic21f9bc64038073a2e97dc906e09ef4bab042047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/19559a4f552ee14a263ae5ca9e7d04e2947efee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/19559a4f552ee14a263ae5ca9e7d04e2947efee4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
