)]}'
{"id":"openstack%2Fcyborg~854336","triplet_id":"openstack%2Fcyborg~master~I074322bdc3116b4a5af8c3588e4a66ddfbca8c78","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-08-29 07:35:07.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-08-29 06:54:17.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I074322bdc3116b4a5af8c3588e4a66ddfbca8c78","subject":"Remove unused code and commnet","status":"MERGED","created":"2022-08-24 05:58:22.000000000","updated":"2022-08-29 07:36:17.000000000","submitted":"2022-08-29 07:35:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854336","meta_rev_id":"e19dfad6b7fef014267848ac1204eaa54beafe1b","_number":854336,"virtual_id_number":854336,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 07:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":2,"date":"2022-08-29 06:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":1,"date":"2022-08-29 06:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 06:39:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 06:52:48.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"f2114a30a8bb5f6e49f89b18dd4ff47229029f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-24 05:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0760fd82938315bc61f21ca58b4bc68d44d44d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 06:39:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31eeef763f9f4c8581b1c8f074e9966c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27bc4095eab040c6ae80c25561efb3a0 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e593070a55f4aabb3df153859b669bb : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2d66cc3dec4cb8854e27551095a8eb : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f703cac8188142018fc4e3174dea5fec : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa5e7b88475d4be0bbdeef9204f17f60 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/075ab0a1d54941ae9918b3f614021088 : SUCCESS in 4m 15s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d1fef8da51ef4586902a858f797fa63f : SUCCESS in 36m 26s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e56176ace204c2bae19ccbda37c1227 : SUCCESS in 25m 31s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/7aa8812c4fc24a22bd77b0386c1a794e : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dbc5a9ca7baccf157aec4a602a5e13b9ee98f3f0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-29 06:54:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8b9dd90aed0d8f4bed97ef096797be7e989db4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 06:54:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcadfa6a31ad13a93954cc5ba23fa764bd9d0934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:35:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64be0b66560e4e62940afca089d4f886\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e901baa14cc4ed6bdf39a2352bc4685 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df0b647d1b541df9e4b907d9452c507 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5987e13cdaf1475fa043d60336435468 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/924b9ef1206e4c88b0a595ca41a7d04a : SUCCESS in 4m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/10c42661d69549929b7647ba1808eec2 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9b3e26e49675b9191b8c27161d0f0ac3abfca954","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:35:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e19dfad6b7fef014267848ac1204eaa54beafe1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:36:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a216ac2593b44e3fb740567dd6530fb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122a2840785c40a1b7f44b7cb279e760 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54291e5a2ebba1730bfc7e381a6b3f15db1612c9","revisions":{"54291e5a2ebba1730bfc7e381a6b3f15db1612c9":{"kind":"REWORK","_number":1,"created":"2022-08-24 05:58:22.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/36/854336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/36/854336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/36/854336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/36/854336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/36/854336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/36/854336/1"}}},"commit":{"parents":[{"commit":"22e7497a926501855d00954ec7612e3b57702977","subject":"Refactor vf attribute generate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/22e7497a926501855d00954ec7612e3b57702977"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-08-24 05:58:02.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-08-24 05:58:02.000000000","tz":480},"subject":"Remove unused code and commnet","message":"Remove unused code and commnet\n\nChange-Id: I074322bdc3116b4a5af8c3588e4a66ddfbca8c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/54291e5a2ebba1730bfc7e381a6b3f15db1612c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/54291e5a2ebba1730bfc7e381a6b3f15db1612c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
