)]}'
{"id":"openstack%2Fcyborg~864826","triplet_id":"openstack%2Fcyborg~master~I5c1790b56fe000239cffdcb213bfc18a919c2e02","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-11-29 06:45:39.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-12-01 07:05:10.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-12-01 07:05:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c1790b56fe000239cffdcb213bfc18a919c2e02","subject":"Add get_one for attributes.","status":"MERGED","created":"2022-11-24 07:11:52.000000000","updated":"2022-12-01 07:06:06.000000000","submitted":"2022-12-01 07:05:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864826","meta_rev_id":"ddd5158d76a41b70a7ad56df87f5c060e6a05001","_number":864826,"virtual_id_number":864826,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 07:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-12-01 06:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2022-12-01 06:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 07:19:41.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-11-24 07:19:41.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2022-11-24 08:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"631c07c3947b21c9add4030f033916d73c7cd00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-24 07:11:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e3926dd79c2a6074e94129dcc9f7f0b476081d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-24 07:44:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"47af014286b62bb9dec8a8bdf18da5c669fe7f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 08:30:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/791ef63ee2cc46f488edda98f4618102\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f92d15defa5342968e123618ca4e6121 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01f7f94316f45c5ac8ef695bf0e5f84 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973106f0859e46da87db9fc9b4949124 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97c61822a5004f6e9c9232643ec12ac7 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f233ce348184dde8fdd71380ec2c1b6 : SUCCESS in 3m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/37ff373c6f934a94a006bf5a3965ddbb : SUCCESS in 23m 41s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c4e311d7020432ab5e4aa8b56c4f5b5 : SUCCESS in 38m 13s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/0d4bafab6ec147d9969d8dd37815217f : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"82c51c2f180c89103cdb3ed81aab0254e8b56c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-24 08:48:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6c060cabc9b031c4185a5cfc66124545840830c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-24 08:56:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dcd0409482f141ee4aadeec8e2714bcd39c0e1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 09:43:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66eff1c5f16f4586a5084200cb867614\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbcba9ec6b5b40619ec1ec09bd3600d9 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc490b3c16f437f8d65d1da19fa1bbf : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237569b4710040c8a9a27633ef371d0f : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b91566c5b7254498a0395bbc28992872 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d5cd04209524570b46a63b0960e9bd8 : SUCCESS in 3m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/677eb74eeee049a68ed4b63383b5f261 : SUCCESS in 38m 51s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebeeac410477423e8f1540b6eb3db912 : SUCCESS in 38m 12s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/535c03d9f33548369ef6f5caff0daee8 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":4},{"id":"9402a074d9d23dd846eaa2cdf8baf1e633e91d34","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-29 06:45:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3040e33d14fe788581dc4bb4bd6f978e14e3f179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-12-01 01:37:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2c887c84a7c6013953a0fedd22345b0ab8d06bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 02:18:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91d2e14d603348f2a30a02bb14da5213\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c6cd61870a4cad8420af4f1a9c7f67 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759487a47fa1401482d22d2b535b0374 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e204a6243514bc1b561ef81d7b3a5f7 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26bc731ff9d94c668058bc1a1478c88e : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7504089a9a594c26bde39652ec1dacbd : SUCCESS in 3m 32s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/146da8fbad3a441fbb30f70ef7942ae2 : SUCCESS in 33m 17s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0c657a5044465fb6dfeb1ecf5a1dc0 : SUCCESS in 33m 06s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/c46300779f5248d1bdea63e259ad410d : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":5},{"id":"7d875e72d6e3f194dcc50d3f477deaf4597885be","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-12-01 06:21:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"458e74394d5a6d6af9376df3cce4531593bd0af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 06:22:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c5c798eea0679a82df0d0d0d9e01c0e7b82b7fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 07:05:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2040f10b689f493dae6b5b04ff364c38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e37ed7db854493b1e62da3f9662651 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2551e78d8b4616b788baa7021d68fa : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/601cd93280744ca7a5aecc90ebe29437 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58a4a89092304aa9bed8979ce68b3bf6 : SUCCESS in 3m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dab40bc0f3d34665a14ac309e8d0b7e8 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":5},{"id":"254356ea1c42fcdb806536f4f6dbc2f205f2622a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 07:05:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ddd5158d76a41b70a7ad56df87f5c060e6a05001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 07:06:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d41a42f0315742da8bc434d066ff37bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd11cb8464746f0803454947a35622f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2eb0be1fa308349daebcf1aedc3e8ead65c15781","revisions":{"14ffa8855b256d46a9e793913b3d8953a10163a2":{"kind":"REWORK","_number":1,"created":"2022-11-24 07:11:52.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/26/864826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/26/864826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/26/864826/1"}}},"commit":{"parents":[{"commit":"efed4e36c6b824a5a5bfd78989a6a1b57cf70743","subject":"Add attribute_list_api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/efed4e36c6b824a5a5bfd78989a6a1b57cf70743"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:11:18.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:11:26.000000000","tz":480},"subject":"Add get_one for attributes.","message":"Add get_one for attributes.\n\nChange-Id: I5c1790b56fe000239cffdcb213bfc18a919c2e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/14ffa8855b256d46a9e793913b3d8953a10163a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/14ffa8855b256d46a9e793913b3d8953a10163a2"}]},"branch":"refs/heads/master"},"3a0641c2d49821c27a621c8673f9c83b7c26f4f7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-24 07:44:44.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/26/864826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/26/864826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/26/864826/2"}}},"commit":{"parents":[{"commit":"247dfb1df057cca7d33e80bdbf9ac28051e58de4","subject":"Add attribute_list_api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/247dfb1df057cca7d33e80bdbf9ac28051e58de4"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:11:18.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:44:44.000000000","tz":0},"subject":"Add get_one for attributes.","message":"Add get_one for attributes.\n\nChange-Id: I5c1790b56fe000239cffdcb213bfc18a919c2e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3a0641c2d49821c27a621c8673f9c83b7c26f4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3a0641c2d49821c27a621c8673f9c83b7c26f4f7"}]},"branch":"refs/heads/master","description":"Rebase"},"dcb962d64479864841adb9180ee6a63df3d42c9a":{"kind":"REWORK","_number":3,"created":"2022-11-24 08:48:41.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/26/864826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/26/864826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/26/864826/3"}}},"commit":{"parents":[{"commit":"247dfb1df057cca7d33e80bdbf9ac28051e58de4","subject":"Add attribute_list_api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/247dfb1df057cca7d33e80bdbf9ac28051e58de4"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:11:18.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 08:48:16.000000000","tz":480},"subject":"Add get_one for attributes.","message":"Add get_one for attributes.\n\nChange-Id: I5c1790b56fe000239cffdcb213bfc18a919c2e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dcb962d64479864841adb9180ee6a63df3d42c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dcb962d64479864841adb9180ee6a63df3d42c9a"}]},"branch":"refs/heads/master"},"108d9dcab4f573cc88450082330a97fbb8ba4b4d":{"kind":"REWORK","_number":4,"created":"2022-11-24 08:56:57.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/26/864826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/26/864826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/26/864826/4"}}},"commit":{"parents":[{"commit":"247dfb1df057cca7d33e80bdbf9ac28051e58de4","subject":"Add attribute_list_api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/247dfb1df057cca7d33e80bdbf9ac28051e58de4"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:11:18.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 08:56:34.000000000","tz":480},"subject":"Add get_one for attributes.","message":"Add get_one for attributes.\n\nChange-Id: I5c1790b56fe000239cffdcb213bfc18a919c2e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/108d9dcab4f573cc88450082330a97fbb8ba4b4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/108d9dcab4f573cc88450082330a97fbb8ba4b4d"}]},"branch":"refs/heads/master"},"2eb0be1fa308349daebcf1aedc3e8ead65c15781":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-12-01 01:37:54.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/26/864826/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/26/864826/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/26/864826/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/26/864826/5"}}},"commit":{"parents":[{"commit":"962ceed37100a2a44d5f64b6ad779b6777bf7b55","subject":"Add attribute_list_api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/962ceed37100a2a44d5f64b6ad779b6777bf7b55"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-24 07:11:18.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-12-01 01:37:54.000000000","tz":0},"subject":"Add get_one for attributes.","message":"Add get_one for attributes.\n\nChange-Id: I5c1790b56fe000239cffdcb213bfc18a919c2e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2eb0be1fa308349daebcf1aedc3e8ead65c15781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2eb0be1fa308349daebcf1aedc3e8ead65c15781"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[]}
