)]}'
{"id":"openstack%2Fcyborg~868963","triplet_id":"openstack%2Fcyborg~master~I94c5331a32edfbc72b3b7fffce7ddbbc123a3274","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-01-01 09:34:33.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2023-01-01 09:59:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94c5331a32edfbc72b3b7fffce7ddbbc123a3274","subject":"Add placement get rp traits api exception ut","status":"MERGED","created":"2023-01-01 03:20:06.000000000","updated":"2023-01-01 10:00:06.000000000","submitted":"2023-01-01 09:59:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868963","meta_rev_id":"cb02c7532ba2957f6907de4d93a9c34d2f27e240","_number":868963,"virtual_id_number":868963,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-01 09:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2023-01-01 09:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-01-01 09:34:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-01 04:06:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-01 08:21:28.000000000","updated_by":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"33d96551f55d00ee361f8c4ff46a9721d3908d62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2023-01-01 03:20:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0456ee2fd12cd1c213029c50c02c48f2d0c68b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 04:06:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa9c58ee25d64f9297305850ebd356de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fdc141288554ab7b0bcc2d230484fcf : FAILURE in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c908282d8c9d41e8847c6d1a821747b8 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a4f2dcbeb54258ba395894ee033dde : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5ddba7f74904c6aa8111a4bc7495b33 : FAILURE in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8a5be8db91641a6ade4b1c30b88bd2e : FAILURE in 4m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5e369edaaf9c4f878f09ae8c342776ad : SUCCESS in 33m 19s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d0063b2c8a04104b9824c3d02251d32 : SUCCESS in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22731fa2e65061dbacb6696f9360e4c75876c0d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2023-01-01 06:32:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e562f9eb1e096ae6af1b578c476aab1237ac9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 06:32:54.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cyborg for 868962,3","accounts_in_message":[],"_revision_number":2},{"id":"ff222a00ab0fa31bbad8179708ad3b7e01c1c012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2023-01-01 06:46:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8435ebbec8581a28e64522045e22c473d93e8122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 07:29:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/640d51bb7dfd42d79d0da3fa4c986d2b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17b7140de4664e3eba9c4ecad067794f : FAILURE in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7db1ccadc0b484088da5ba62f051028 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87acee6abd447a5a9a4c952a213e238 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7555c3dbd89d443a8a14cae89541f7a1 : FAILURE in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2df7e11d2b4a4ed69ae7e29de396c697 : FAILURE in 3m 23s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c0edb0b480fa43289abf196c640e4b95 : SUCCESS in 19m 15s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c42ace7bfad4e859207d9e91e2b0b6f : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5812401a29ea75e671ced21660c19505a9e30f30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2023-01-01 07:31:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1891edc1a73fcd50278211c648dc268456ada587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 08:12:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cfcf19670994958920a46ae113949c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55de21bda6f14f15ac0aa7d68d3c280f : FAILURE in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5ae0ee7fa541bda2fe342f08116757 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b0146a0e9d4d9883b5ab0c1dd13d98 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecd27653af5d4e6c8747ef156b229842 : FAILURE in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf175a24f0594c33b2a92dd688dd7644 : FAILURE in 3m 23s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/034f3ccf3af64369b0fc7e836bfa0f17 : SUCCESS in 16m 23s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2acc6b366c66493999dc01680cc8a854 : SUCCESS in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2d645fc8d2fae947ef8e48b43ceaf42f45ba4543","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2023-01-01 08:20:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e3abbd30e3f15115fc079cd9ae527d1e8b536c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 08:58:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6074424b909c4307a9635f19d8d4ac7f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b80639d27d874eef805b262304946815 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fcbc1b4731841c18a3c2a1ea8b965a8 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5429761cec424795eb6d13108a443a : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85f4a40f72284ee9b425e23c2dcc5829 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d83f752cda6c4a13bd391ac0304ed6b9 : SUCCESS in 3m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bd4e32066c65439084d14c5e32d11d4a : SUCCESS in 31m 13s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ef6e71826894b138a15e965a04ac278 : SUCCESS in 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db929b2e6bc52d18914fbd779f5be7e4d369156c","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-01-01 09:34:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ca9f0466b86921c6f744aaad1f974681426fcd6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 09:35:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d9cb8b05cda0a2e2378953b8564f3eeda6c5f4ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 09:59:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de9fe91442f34a8ea7ff08172ce881f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c4375160244a139eb6bf9e468894ac : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a9fec1d2984480bf1c44296db9e011 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8674d0809bc64bb280a8faaedc546ed4 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53dacbbb24cf424895701b1026f8bad1 : SUCCESS in 3m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e34e82563b014a70af34f7cb2e31a0bb : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":5},{"id":"bf99013b38fe82baa9d98cfda6d06c6cc15aaeb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 09:59:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"cb02c7532ba2957f6907de4d93a9c34d2f27e240","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 10:00:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb358d75d86a48ec870ed0f292c383f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/894379b47e72482c831461854fda68aa : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cf70b79bdf560dcc57c0cc880942f39ca665f7de","revisions":{"83375bd96508ed036c805726545366a2a7379052":{"kind":"REWORK","_number":1,"created":"2023-01-01 03:20:06.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/63/868963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/868963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/868963/1"}}},"commit":{"parents":[{"commit":"89d573936610a391ac1204524fbfabbcdb5937db","subject":"Add placement get rp traits api ut","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/89d573936610a391ac1204524fbfabbcdb5937db"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 03:19:43.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 03:19:43.000000000","tz":480},"subject":"Add placement get rp traits api exception ut","message":"Add placement get rp traits api exception ut\n\nChange-Id: I94c5331a32edfbc72b3b7fffce7ddbbc123a3274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/83375bd96508ed036c805726545366a2a7379052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/83375bd96508ed036c805726545366a2a7379052"}]},"branch":"refs/heads/master"},"55d682ac65a74852b543814187edbc5db3a54261":{"kind":"REWORK","_number":2,"created":"2023-01-01 06:32:24.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/63/868963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/868963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/868963/2"}}},"commit":{"parents":[{"commit":"89d573936610a391ac1204524fbfabbcdb5937db","subject":"Add placement get rp traits api ut","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/89d573936610a391ac1204524fbfabbcdb5937db"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 03:19:43.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 06:32:02.000000000","tz":480},"subject":"Add placement get rp traits api exception ut","message":"Add placement get rp traits api exception ut\n\nChange-Id: I94c5331a32edfbc72b3b7fffce7ddbbc123a3274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/55d682ac65a74852b543814187edbc5db3a54261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/55d682ac65a74852b543814187edbc5db3a54261"}]},"branch":"refs/heads/master"},"c8e6369ecd7862cb41ac2a0138693e5a8dab96a4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-01 06:46:54.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/63/868963/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/868963/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/868963/3"}}},"commit":{"parents":[{"commit":"7fd543ffac3301d4bcae26672e19ba9254cb53bb","subject":"Add placement get rp traits api ut","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7fd543ffac3301d4bcae26672e19ba9254cb53bb"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 03:19:43.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 06:46:01.000000000","tz":480},"subject":"Add placement get rp traits api exception ut","message":"Add placement get rp traits api exception ut\n\nChange-Id: I94c5331a32edfbc72b3b7fffce7ddbbc123a3274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c8e6369ecd7862cb41ac2a0138693e5a8dab96a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c8e6369ecd7862cb41ac2a0138693e5a8dab96a4"}]},"branch":"refs/heads/master"},"44bf19dd55f093fc161bc5c601a3277923e38567":{"kind":"REWORK","_number":4,"created":"2023-01-01 07:31:32.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/63/868963/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/868963/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/868963/4"}}},"commit":{"parents":[{"commit":"5c64314a93a8fc12bbb071912cc9dda8c0bac11e","subject":"Add placement get rp traits api ut","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5c64314a93a8fc12bbb071912cc9dda8c0bac11e"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 03:19:43.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 07:20:40.000000000","tz":480},"subject":"Add placement get rp traits api exception ut","message":"Add placement get rp traits api exception ut\n\nChange-Id: I94c5331a32edfbc72b3b7fffce7ddbbc123a3274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/44bf19dd55f093fc161bc5c601a3277923e38567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/44bf19dd55f093fc161bc5c601a3277923e38567"}]},"branch":"refs/heads/master"},"cf70b79bdf560dcc57c0cc880942f39ca665f7de":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-01 08:20:27.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/63/868963/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/63/868963/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/63/868963/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/63/868963/5"}}},"commit":{"parents":[{"commit":"8554e50bc4975f25fdf4eeb0e0c5058b212f0f1f","subject":"Add placement get rp traits api ut","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8554e50bc4975f25fdf4eeb0e0c5058b212f0f1f"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 03:19:43.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-01-01 08:13:54.000000000","tz":480},"subject":"Add placement get rp traits api exception ut","message":"Add placement get rp traits api exception ut\n\nChange-Id: I94c5331a32edfbc72b3b7fffce7ddbbc123a3274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf70b79bdf560dcc57c0cc880942f39ca665f7de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cf70b79bdf560dcc57c0cc880942f39ca665f7de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[]}
