)]}'
{"id":"openstack%2Fcyborg~868970","triplet_id":"openstack%2Fcyborg~master~I87567ce725da97332ab5534346662d05b556e5c7","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2023-01-01 08:13:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I87567ce725da97332ab5534346662d05b556e5c7","subject":"add attributes api ut file","status":"ABANDONED","created":"2023-01-01 07:00:52.000000000","updated":"2023-01-01 08:13:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"37195c9e48aeac548648d144baf39e74becb8b39","_number":868970,"virtual_id_number":868970,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-01 07:21:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-01 07:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"947a06df4f1e69099f6ba21eea279fefd2b7b475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2023-01-01 07:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e85b8ce6b1f8061b227c562db394527ca42d73cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 07:21:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c64e91b2e384a39835e032b8cf9936f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31dba445bb19499da33f534295f3701c : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313e209fd6ea43d7a6d6e9f416660977 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26da9269f1b74f779ebca4fdb8141dc1 : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a25aa98bd0084842a5371bbe5dc8a53c : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe1cafb7c66b458789006b6b9f06be22 : SUCCESS in 3m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8fdec0af5a1d40128a33b8508c2e82c0 : SUCCESS in 18m 37s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29b9d964803349b68cc7f80a00ce8748 : SUCCESS in 19m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37195c9e48aeac548648d144baf39e74becb8b39","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2023-01-01 08:13:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ed8f674e3d8143ead10a3578460349784fd0558","revisions":{"9ed8f674e3d8143ead10a3578460349784fd0558":{"kind":"REWORK","_number":1,"created":"2023-01-01 07:00:52.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/70/868970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/70/868970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/868970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/868970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/70/868970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/70/868970/1"}}},"commit":{"parents":[{"commit":"cb1331a8feab5e2b8ab65fd70bf9da33ba754586","subject":"Improve the api-ref docs for v1 version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cb1331a8feab5e2b8ab65fd70bf9da33ba754586"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2023-01-01 06:57:48.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2023-01-01 06:57:48.000000000","tz":480},"subject":"add attributes api ut file","message":"add attributes api ut file\n\nChange-Id: I87567ce725da97332ab5534346662d05b556e5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9ed8f674e3d8143ead10a3578460349784fd0558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9ed8f674e3d8143ead10a3578460349784fd0558"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
