)]}'
{"id":"openstack%2Fcyborg~871161","triplet_id":"openstack%2Fcyborg~master~I1a0258766757ad3baa4422ed4c51632a4ca0330d","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-05-17 00:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"20566":{"account":{"_account_id":20566,"name":"BaiLinZhang","email":"bailin.zhang@hotmail.com","username":"bailin.zhang"},"last_update":"2023-05-17 01:18:29.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-05-17 01:18:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a0258766757ad3baa4422ed4c51632a4ca0330d","subject":"Use new get_rpc_client API from oslo.messaging","status":"MERGED","created":"2023-01-19 20:32:27.000000000","updated":"2023-05-17 01:19:26.000000000","submitted":"2023-05-17 01:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871161","meta_rev_id":"0fb1827cd116d111d0dc4e2119a755d4498558da","_number":871161,"virtual_id_number":871161,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 01:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2023-05-17 00:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2023-01-28 01:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-05-17 00:59:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20566,"name":"BaiLinZhang","email":"bailin.zhang@hotmail.com","username":"bailin.zhang"}],"reviewers":{"REVIEWER":[{"_account_id":20566,"name":"BaiLinZhang","email":"bailin.zhang@hotmail.com","username":"bailin.zhang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 21:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-28 01:07:25.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2023-02-12 14:29:39.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2023-02-12 14:29:39.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":20566,"name":"BaiLinZhang","email":"bailin.zhang@hotmail.com","username":"bailin.zhang"},"state":"REVIEWER"}],"messages":[{"id":"668d9337745b08b4705247112b243605495880e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-19 20:32:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a4a286c3397958dcefa6daa66f908751b4f623d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 21:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7347357d6ea7498f8164ff5af04364d4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4953c841c5814c849d09542b01bad6da : SUCCESS in 4m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a27280e66214140bd2a08c77e7fe13b : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a4a885b0b94ec1b7167887965f949e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e20bb5b24ac84489b815640039875fcf : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42ae835dc78b4071bd03f6f5f50238cd : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c9c207518c04de28a44c5b9a3f724cf : SUCCESS in 4m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0ddd5dee1d164a078a1eb68844addb76 : SUCCESS in 20m 41s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a38e64d1d3a4106a4b12133b679a1d9 : SUCCESS in 28m 10s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/24fb1d8288cd43fb8ee6884038c49a74 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7c8173ba3a7d60bb307008546975b3dbe5fcf95b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2023-01-28 01:07:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"008ebd8d0971783948722cfa8086db8d755e3e50","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-05-17 00:59:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"253a3eabc70e3602c4ea84238d4cd4bd26fdae43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 01:00:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f172662d0a0cff936ed64c37f04266aabf59956f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 01:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ffaf1ad27343c98ea8bcd7d0c2b7d5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0af3e57a04fe4de4bc354b0ed261abf5 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde56000104e433b8bab9dad94867e80 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce0fa9504c74df09f7f6def1a2b2840 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8283b8baae084c7a87de86d7fcb84a36 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baba3dacb92a4b3fb9ebc338dad2bfcf : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a23f73ae61a8486a96fe4a4d63a4e51e : SUCCESS in 3m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/df49d650e5b7439c90314dbc7e938158 : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9e8310b94840b08ba6e9f65724991b166942a7e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 01:18:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0fb1827cd116d111d0dc4e2119a755d4498558da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 01:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c609302d2af4f5fac163bc7138aef9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0893b0e6194f28be17f1bd96191cf9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6491be0d196d5f1bb3eb41e02157463a3fe7469d","revisions":{"6491be0d196d5f1bb3eb41e02157463a3fe7469d":{"kind":"REWORK","_number":1,"created":"2023-01-19 20:32:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/61/871161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/61/871161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/61/871161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/61/871161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/61/871161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/61/871161/1"}}},"commit":{"parents":[{"commit":"eafcc2fc64df6d8c8c2e3b709e9d4e7d5e7be9b7","subject":"Fix the wrong use of variable assertions in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/eafcc2fc64df6d8c8c2e3b709e9d4e7d5e7be9b7"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-19 20:32:07.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-19 20:32:13.000000000","tz":0},"subject":"Use new get_rpc_client API from oslo.messaging","message":"Use new get_rpc_client API from oslo.messaging\n\nUse the new API that is consistent with\nthe existing API instead of instantiating the client\nclass directly.\n\nThis was introduced in release 14.1.0 here [1] and\nadded into oslo.messaging here [2]\n\n[1] https://review.opendev.org/c/openstack/requirements/+/869340\n[2] https://review.opendev.org/c/openstack/oslo.messaging/+/862419\n\nChange-Id: I1a0258766757ad3baa4422ed4c51632a4ca0330d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6491be0d196d5f1bb3eb41e02157463a3fe7469d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6491be0d196d5f1bb3eb41e02157463a3fe7469d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
