)]}'
{"id":"openstack%2Fcyborg~883651","triplet_id":"openstack%2Fcyborg~master~I7a97cecff3bdfa1ee80bb1d2ee5f8a9e9d1448cf","project":"openstack/cyborg","branch":"master","topic":"bug/2020179","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2024-01-26 05:53:17.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}},"hashtags":[],"change_id":"I7a97cecff3bdfa1ee80bb1d2ee5f8a9e9d1448cf","subject":"Fix delete attribute api bug","status":"MERGED","created":"2023-05-19 10:09:10.000000000","updated":"2024-01-26 06:16:43.000000000","submitted":"2024-01-26 06:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883651-bug/2020179","meta_rev_id":"6eaeb1f5f0d9a737046082505f95b492991d9916","_number":883651,"virtual_id_number":883651,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-26 06:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2024-01-26 05:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2024-01-26 05:53:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 10:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"153f149bc079d53f29e9fa92a909235978f7ead7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-05-19 10:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2836358a20505f1af55bc83047532f9ab1e039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 10:44:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b3ba0e309714e599e7a293e9c6f906e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b468fa3235b4f21a98516c8f0d8eb0b : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a2189634b04acc823f139994c56948 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61056d0dc6674f9289c0fb064841275a : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d26c9307de74f4dbd8c54cc67841031 : FAILURE in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d06fc6ed0251402c94b9fb41a509e4a2 : FAILURE in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77ea7996da4e443cb6b76e25456121de : FAILURE in 3m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8c9cef6568434ccdb6ab784fed8cf6bc : SUCCESS in 34m 35s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93e3dff634f44b8ca3d48625f5775aef : SUCCESS in 27m 08s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/f99e936109244b9a8c1f931242fe8a72 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"75c6b4570045f57db8e84f78c86314354f5d1e98","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-01-26 01:00:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84555180da4d49ba0d9e604418e9885e9db04720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 01:20:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf0504923c74d688f50a1fdbed333d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/374eea9ce3034d78a76d031d0e8f9518 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc72eaf1ed694a80891cc9768981648a : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08b2d3e3c544858b5388e31478dca7e : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de3c543095c4c69ae8b02acf16e74eb : FAILURE in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68cd39fc0a084219a6f55dd12c7132ad : FAILURE in 3m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fb8d2dd2e5554cb6bb385e414ec112d5 : SUCCESS in 17m 57s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f28be8c4abb4e38b95925de826781ca : FAILURE in 18m 38s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/ab3641cf054b4e20bb68ef20f72986d6 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8e47a6264db49f254ff14558e586fc3c2a8f9219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-01-26 02:58:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b9f43969d0626b9a858e3c78eef20fcef8c25677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 03:53:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a27331c31e7460b9a3b87a1cba6d339\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f10e9ff3c7944d69c0726320a1a7c9d : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b09d39569b2c408094acabf766a1704c : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95258a7518644a72b748f54a38aa2478 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa287296e54a456297c17675d08808a2 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/532006b29407479a88b2ec5759259968 : SUCCESS in 4m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ffa8de7dba12423ea246bdc64bfbfa81 : SUCCESS in 36m 32s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cb99236c00d4a5ba84c1b3e9b5205ec : FAILURE in 35m 03s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/1ef2ef5aa481482089e2edc57af7a8c0 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3b75d86271fa98c9d9eac794571b74fe52e70b29","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-01-26 05:53:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97932cbadc55f9edb18188f1c9d79d4e95703bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 05:53:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d6c45274baee0270e5af53129bc8d5c5f2fcbfb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 06:15:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da25ad7d822e484685ef25f7fce20f69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff4ea72882246398228c90d7ea6ae2e : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7717234679d847889447ddfae9b56db5 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e90faf8991841629aa8f40ff3aca7b5 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e2318d63cab4432badb9d75ae977842 : SUCCESS in 3m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/af57beb6469c492ebee52187b04c8acb : SUCCESS in 20m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b13daa30330f3d708088c994052fa6dccee8486f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 06:15:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6eaeb1f5f0d9a737046082505f95b492991d9916","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 06:16:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfd4f030bbca4e1a8f08e434bbbca329\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcb96ddf9d44b01b358227c124f6bf7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df85db5677776da4f6f4f80b5d75aa9e2f80519a","revisions":{"51df465d1baf20879680f60bdc7471dababdd08d":{"kind":"REWORK","_number":1,"created":"2023-05-19 10:09:10.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/51/883651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/51/883651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/51/883651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/51/883651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/51/883651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/51/883651/1"}}},"commit":{"parents":[{"commit":"b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e","subject":"improve doc format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-05-19 10:06:42.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-05-19 10:09:02.000000000","tz":480},"subject":"Fix delete attribute api bug","message":"Fix delete attribute api bug\n\nCloses-Bug: #2020179\nChange-Id: I7a97cecff3bdfa1ee80bb1d2ee5f8a9e9d1448cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/51df465d1baf20879680f60bdc7471dababdd08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/51df465d1baf20879680f60bdc7471dababdd08d"}]},"branch":"refs/heads/master"},"df85db5677776da4f6f4f80b5d75aa9e2f80519a":{"kind":"REWORK","_number":2,"created":"2024-01-26 02:58:28.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/51/883651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/51/883651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/51/883651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/51/883651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/51/883651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/51/883651/2"}}},"commit":{"parents":[{"commit":"b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e","subject":"improve doc format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-05-19 10:06:42.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2024-01-26 02:58:06.000000000","tz":480},"subject":"Fix delete attribute api bug","message":"Fix delete attribute api bug\n\nCloses-Bug: #2020179\nChange-Id: I7a97cecff3bdfa1ee80bb1d2ee5f8a9e9d1448cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/df85db5677776da4f6f4f80b5d75aa9e2f80519a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/df85db5677776da4f6f4f80b5d75aa9e2f80519a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
