)]}'
{"id":"openstack%2Fcyborg~886172","triplet_id":"openstack%2Fcyborg~master~I2ed16ab3e779715cd7f7453bf61cab7069adecc6","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"35819":{"account":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"last_update":"2023-06-16 03:20:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ed16ab3e779715cd7f7453bf61cab7069adecc6","subject":"Add accelerator.common.utils test","status":"MERGED","created":"2023-06-15 12:06:48.000000000","updated":"2023-06-16 03:21:20.000000000","submitted":"2023-06-16 03:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886172","meta_rev_id":"7045e563461b57fac73355df104a8e0778c168f2","_number":886172,"virtual_id_number":886172,"owner":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-16 03:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2023-06-16 02:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-06-16 02:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 12:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 02:47:52.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"b6c2701ea853a1e329cf0e8c5e3a6d57a389baa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"date":"2023-06-15 12:06:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fee3353b1658635679e680335fc540ca72a18058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 12:50:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/594a57ad04d44d0591115ea9c8e36a64\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/292a25a0a5724713bcdcb2790d782f63 : FAILURE in 33m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1341dab5e84b47b98dea57cf368849 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0780177f034a488d8f92ebe8c16aec4b : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/138ab30b1c2045118816094326d08549 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38dcf4f431654e1ba92b89942d8ad5be : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c0cf74269234fe1a035aadbf2541754 : SUCCESS in 4m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b95342d831b942d48bbf4b71fa17ed7a : SUCCESS in 23m 43s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e254035514446c1ad6f092fd6b9cfca : SUCCESS in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e7d9a98367fafb2bf6fe40d1fa11ecc12bd6819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"date":"2023-06-16 01:08:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa33c1013ad6ad1db6a57da9eb3e40be39c4c37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 01:54:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ace20bc8ad134343b0422dafd016e1c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed0e80a40bac452eb064d92b992deb42 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3bedec20ea49adb58bb8fc6f9650d2 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af95ce9cefe14e38b1c4dcc64af0a37b : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1185d1ca75f4c33ad0d25b8e14bf1e7 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e33c79b9e53a41eaa242dba8a1053c29 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6931ae41ab51433a88731bd3fc67239f : SUCCESS in 3m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8f6ff38f7ee2499f8466d34d19de1673 : SUCCESS in 21m 25s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f29553db665418cad6f6ebdc4afb3ac : SUCCESS in 38m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1a3d2ab20752242cc97c2cb09d9cd8a0e850782","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-06-16 02:47:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3fd4bc73277cc83fa6ad24d08d66aa32b98a815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 02:48:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ba51dc1eb3ed11cebaf778f529a9010d594c41a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 03:20:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ee886baba94e1c990f57799f80c15c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d291ca636f4fa7885eea676a4d1ade : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b333b7c8f2aa4c16a010bad349160ac8 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b943c9237f4179972bb1371456d8b3 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21319d87eff64b4fab83ef0de7df26b9 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3204348186ab483d9fac9bd1e315469e : SUCCESS in 4m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ce530d7255f14075b87555e00ce6d342 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f390132842278dc1e2656e9c1dc4d082dad00612","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 03:20:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7045e563461b57fac73355df104a8e0778c168f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 03:21:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e32eb052f6c44a338a1ecd2cf396c603\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ce2f9735b2490db421ee713f39c6ac : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0e7892d2ff1871fd7b3078c05d0dee9c0666a5d","revisions":{"20e10b96a5e98448f9708af1875d564adad92463":{"kind":"REWORK","_number":1,"created":"2023-06-15 12:06:48.000000000","uploader":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"ref":"refs/changes/72/886172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/72/886172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/72/886172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/72/886172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/72/886172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/72/886172/1"}}},"commit":{"parents":[{"commit":"b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e","subject":"improve doc format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e"}]}],"author":{"name":"zhuyawei","email":"zhuyawei@inspur.com","date":"2023-06-15 11:50:41.000000000","tz":480},"committer":{"name":"zhuyawei","email":"zhuyawei@inspur.com","date":"2023-06-15 11:50:41.000000000","tz":480},"subject":"Add accelerator.common.utils test","message":"Add accelerator.common.utils test\n\nThis patch is proposed to test accelerator.common.utils scripts.\n\nChange-Id: I2ed16ab3e779715cd7f7453bf61cab7069adecc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/20e10b96a5e98448f9708af1875d564adad92463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/20e10b96a5e98448f9708af1875d564adad92463"}]},"branch":"refs/heads/master"},"b0e7892d2ff1871fd7b3078c05d0dee9c0666a5d":{"kind":"REWORK","_number":2,"created":"2023-06-16 01:08:15.000000000","uploader":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"ref":"refs/changes/72/886172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/72/886172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/72/886172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/72/886172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/72/886172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/72/886172/2"}}},"commit":{"parents":[{"commit":"b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e","subject":"improve doc format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b4ca84cd90ab37ebf2c73cb57d6b0ed29c23ea5e"}]}],"author":{"name":"zhuyawei","email":"zhuyawei@inspur.com","date":"2023-06-15 11:50:41.000000000","tz":480},"committer":{"name":"zhuyawei","email":"zhuyawei@inspur.com","date":"2023-06-16 01:06:50.000000000","tz":480},"subject":"Add accelerator.common.utils test","message":"Add accelerator.common.utils test\n\nThis patch is proposed to test accelerator.common.utils scripts.\n\nChange-Id: I2ed16ab3e779715cd7f7453bf61cab7069adecc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b0e7892d2ff1871fd7b3078c05d0dee9c0666a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b0e7892d2ff1871fd7b3078c05d0dee9c0666a5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
