)]}'
{"id":"openstack%2Fcyborg~889700","triplet_id":"openstack%2Fcyborg~master~I069b555489bdd35834ca6cbc57827dc50027b1cd","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"I069b555489bdd35834ca6cbc57827dc50027b1cd","subject":"Change the log level and content","status":"MERGED","created":"2023-07-26 11:37:41.000000000","updated":"2023-07-27 07:39:09.000000000","submitted":"2023-07-27 07:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889700","meta_rev_id":"c73d1b434cbf9749484312428cc42ad980584841","_number":889700,"virtual_id_number":889700,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-27 07:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2023-07-27 07:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-07-27 07:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-26 12:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84711ee10ff20f08fdb054f3d40037a6226f97df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-07-26 11:37:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f121f62235dcd2b8ca4b9089d89f5b3879728681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 12:24:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a077ab7112e42d99935eb6b25381c2f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e91287627214e57a29a8fe894068627 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca16edaf0944bc3801cc3852e444b05 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a115caaeea484bd09df3706115b749c4 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/364164f3f454473b9e4a1af63ef7628a : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69c12fe90ce74a2aa8a3ef4aa93829e6 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c51ac9dffa848e49e2e78f4d608376e : SUCCESS in 3m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d3be17029d96496c84a25854c4337598 : SUCCESS in 39m 10s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c27906940fa140cda0534f2c79c7fc86 : SUCCESS in 22m 06s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/2c52ae6ef6284115a043ad065c7869a4 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"efc38f9b10be33d7d7ea78f8ae0dab4118b20263","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-07-27 07:08:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1e207501fabfd5d34b1a0aa4718ad28f2ead0d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:09:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca17fd6148bc2748e1e97fb2d5d17b1805e35a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:38:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec97eab92694f80919fee9192ac2928\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1275009027f440a18cac1d05268f90cb : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3bf853c0434583b2a9ba3085a6fcae : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72f830e8d3194936a0ef0319b4831964 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3873452939854ef9bcd3cb0dbd2728b6 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40dcf50a743040a2931c0cce56fc2652 : SUCCESS in 4m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d2b02062c4dc497dbeba85e88abf31ce : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f2174149aac95cdda1c02f6f37f089f8428d82a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:38:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c73d1b434cbf9749484312428cc42ad980584841","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:39:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8172766440ca4f3b9f682b84a2f4f8f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fbe38138884ce9a35e0892eafed4bb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a34fe93ce933800ff2d64480c9f266b2b4ce5d31","revisions":{"a34fe93ce933800ff2d64480c9f266b2b4ce5d31":{"kind":"REWORK","_number":1,"created":"2023-07-26 11:37:41.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/00/889700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/00/889700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/00/889700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/00/889700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/00/889700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/00/889700/1"}}},"commit":{"parents":[{"commit":"4d329240f95fbb931e8b976f70f0f5b6069f8196","subject":"Merge \"Add pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4d329240f95fbb931e8b976f70f0f5b6069f8196"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-07-26 11:37:22.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-07-26 11:37:22.000000000","tz":480},"subject":"Change the log level and content","message":"Change the log level and content\n\nChange-Id: I069b555489bdd35834ca6cbc57827dc50027b1cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a34fe93ce933800ff2d64480c9f266b2b4ce5d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a34fe93ce933800ff2d64480c9f266b2b4ce5d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
