)]}'
{"id":"openstack%2Fcyborg~906811","triplet_id":"openstack%2Fcyborg~master~I859bb0097fd0ace3452ca829eeee2c71f80491f2","project":"openstack/cyborg","branch":"master","hashtags":[],"change_id":"I859bb0097fd0ace3452ca829eeee2c71f80491f2","subject":"remove max version desc","status":"ABANDONED","created":"2024-01-26 07:13:27.000000000","updated":"2025-08-19 01:25:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44e7e19f91bd6868e770cfa63f46bc361cbac308","_number":906811,"virtual_id_number":906811,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-26 07:34:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 07:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6dabde1a036e23cb1a2d1f5163c0e893eb6e527c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-01-26 07:13:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9999716e7edf1c08ba3c1dc02ebbf55ce3da18ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 07:34:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41e4033de51d48e9bf9c88fb02e0b698\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a5050b7c1cc94e018902a899b822cf0d : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5c7ed2e77344fca7db9de5cebeba6f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a0d76fc0e84669a49e4aece8256521 : SUCCESS in 3m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a0f8ea8d50ab46a487e557f53c1b9c35 : SUCCESS in 18m 38s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/134988dd931f44e0a4e55da2bd1ece8f : FAILURE in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44e7e19f91bd6868e770cfa63f46bc361cbac308","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2025-08-19 01:25:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ea89a060384186cf97af969667bf1807158640a","revisions":{"2ea89a060384186cf97af969667bf1807158640a":{"kind":"REWORK","_number":1,"created":"2024-01-26 07:13:27.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/11/906811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/906811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/906811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/906811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/906811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/906811/1"}}},"commit":{"parents":[{"commit":"127cea02e86540b841dd8ff6bfb83ed99c2b9629","subject":"docs: add request header example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/127cea02e86540b841dd8ff6bfb83ed99c2b9629"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2024-01-26 06:49:09.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2024-01-26 06:49:09.000000000","tz":480},"subject":"remove max version desc","message":"remove max version desc\n\nChange-Id: I859bb0097fd0ace3452ca829eeee2c71f80491f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2ea89a060384186cf97af969667bf1807158640a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2ea89a060384186cf97af969667bf1807158640a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
