)]}'
{"id":"openstack%2Fcyborg~928711","triplet_id":"openstack%2Fcyborg~master~Iafbd5470e55e10ffb6bcddd232a698eb042d7eed","project":"openstack/cyborg","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2024-09-10 06:10:07.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-18 07:08:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafbd5470e55e10ffb6bcddd232a698eb042d7eed","subject":"Drop system scope from RBAC","status":"MERGED","created":"2024-09-09 21:38:31.000000000","updated":"2024-09-18 07:09:55.000000000","submitted":"2024-09-18 07:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928711-secure-rbac","meta_rev_id":"ddab2ebdeeb959fac54c14d9ffe0b5a050101c6d","_number":928711,"virtual_id_number":928711,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-18 07:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2024-09-18 06:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2024-09-18 06:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 22:10:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 23:31:31.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2024-09-10 04:23:29.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"CC"},{"updated":"2024-09-10 04:23:29.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"18b821a3a43f718eddfcb4fbf74f5a198099258b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-09 21:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9db9091b8c22af3c5c092b691dab0f8a5fbf2e93","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-09 21:38:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"26dd0c2f0cddb2605fbc055a2abcdecf75688bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:10:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58c45bd313e54932ad7d17990980f3d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce7a6e1e09447be9ca00b6d81e31d60 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e7f09fa9ca444e8cd019b325b0b761 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f009d665b6c4fcb9f4f6737ecae5961 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3a43ca75d2746dda5401727d6e64137 : FAILURE in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e9267ecff994511a772e3a6600e7f99 : FAILURE in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08bcc7215f4245acac525ede5bdb16f6 : FAILURE in 3m 32s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fbb150500c344774af80aef7d8cad20f : FAILURE in 30m 55s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a0811c7a1a44a3cb784a420b1345750 : FAILURE in 16m 29s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/102d24a07a9f471fac985db7a5e8ac93 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c4059b4e87fc87f768a4c5f02f94fb5848cccfb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 04:00:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fe509428614a2d25d4b08974ace2f42812267099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 04:57:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78b47e823bd148ed9a55707fe3b79bb9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/085cf0b94dae4b95b90895aa5c6ddbc4 : TIMED_OUT in 50m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfa593462104e02a8f24d6f1d942716 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ceb2db09aa641a19fd9bc471b2ddbb6 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abe534b3970a4636b80bbece54a2aa33 : TIMED_OUT in 40m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0a84ea9ab1b4be2b99afed5e680906d : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f54870f0072f43b19f45eeb12ef258e6 : TIMED_OUT in 40m 45s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3ce17d0a6b3d4cb4a6d1a37e906de524 : SUCCESS in 18m 47s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9e6365562c7404ab331918a6874d2f4 : FAILURE in 21m 53s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/e7bebbc7d454431688c4671f5b2bf575 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f334a5d613657cf3762d10aa64d524daaf8b5c6c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 05:01:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"443333fda95fc40ef0e732aa944fcd2fba40531c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 05:53:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/701c226f3c984f14bdc2d8cba5b63901\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96e038cd115d45a1b4e2413f81eaaa1e : TIMED_OUT in 50m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58fe065b50d4326b61d6df895cb2a59 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29f6f894a68489686840d0e20c74dd0 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9561cbb1f30448da44978937763ead2 : TIMED_OUT in 40m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c31a587ccea49cc9d44be2063f3ae47 : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/277693a0179841a692c832937b220988 : TIMED_OUT in 40m 50s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b395172cf36d4479be019e7178fde622 : SUCCESS in 28m 45s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5c005ce429340bc97942097b8940bed : FAILURE in 27m 27s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/b4578993d61d425ea077b4ad727d997f : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"cbbf2f66b80caa5c16a67b0eb04f689de866ea3f","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-09-10 06:10:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c49ae7f890b6c335133779fc2699a7515b5e7c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 07:04:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e7fa7cc4b7d46a282de35e6ee52b49d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f75f2e86089e4a0da94f4c0f1577e18b : TIMED_OUT in 50m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bee8ffcc3184bbc8bdcc970e1c4a073 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbb95393a83e40838b5cdaf2bbd381b9 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a440e1f57eb441159ee0f9501df4a3e4 : TIMED_OUT in 40m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5ba3bd8d32f4b4aa1d873473f51e250 : TIMED_OUT in 40m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74259404ab0e4d18991ebeb71fda7d09 : TIMED_OUT in 40m 44s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cee3ae795d784559a3030fe3e9c767b2 : SUCCESS in 26m 31s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/704930cc40a946809a4dab185cbc00d4 : FAILURE in 28m 28s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/3b8b9218a0bc4e4a8c53e9198727a0d1 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3c53c6e5971446f1b334e077097ddf6e6f439317","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-09-10 08:58:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46c1b0a3d00f392d585fc72415a7defd1b09b590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 09:51:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c8b1932a1a343d28de60cf9944c02bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b1d6fbea3604dafb760e060921f00f6 : TIMED_OUT in 50m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3dbfdd1f11145d1b7ee55e94d7f46b9 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e324a84f4b411aa4f68c6839f90023 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba3572ea1a3245fc99278a2f12698755 : TIMED_OUT in 40m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7184c7b09e6249d79f3e5c4f88b3b3a0 : TIMED_OUT in 40m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a4b15524a9a4e1ca117e34606ea6edc : TIMED_OUT in 40m 51s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/828664630cd24889a2d3c0909cd83ae6 : SUCCESS in 29m 57s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96ec5a0bbf914d969652cdb420ca497b : FAILURE in 28m 20s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/645c2867254141e58b70076bc27839f7 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"973a1e19056595818f051da468ce66f5c724c728","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 14:59:48.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e04370e8f6d815271c57f381b75bffe2c085807a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 15:58:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"30f0c908c4ae48096dc0c90de0de904b00deff3c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 15:59:42.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"4269aa297068c01a4aa671fe93153f4e225d3725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:04:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ba126239a584d15a227e569928425de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52baab21ec7e46abb3562fdff4fd7070 : TIMED_OUT in 50m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5292a1d5b64c41dbaa7f2bdaf17f6200 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243d866b73f241a9bdb917d4fbb16847 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4739c6502814f3981908da000a3fbe7 : TIMED_OUT in 40m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c991d7955492465081e7b85a84c7d4ed : TIMED_OUT in 40m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb77a01b8e274838a3326dfe93870e50 : TIMED_OUT in 40m 55s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b6fd54b8d289497d871fc47603a4e3b2 : SUCCESS in 19m 36s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/239f338600ba44adadeea324e48894e8 : FAILURE in 31m 45s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/fb8742c2e0e04492a8d98cd666c31c8d : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"10f1329e72886ce047e093687352a586718cb46e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 19:08:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10611bb9624efa286d822f86986b5d8f6fc13e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-13 22:48:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6e591cd8b70ff3601e15a77944782aaa228eb524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-13 22:49:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eb86e2ea03dbdc931a76c1338061c739ed9125e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 23:42:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bda6bfd98a1429da6e178172de0c3c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e54f1f74f6f34f978243164ae98e6269 : TIMED_OUT in 50m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e955df638d418baf22b2b9e78c2204 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35082014467a4c59bed77adc08630f36 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5824554ffd604d238900c59bcd691298 : TIMED_OUT in 40m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8ffc67209b94025b585f9867f4a5d0f : TIMED_OUT in 41m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1dc031ab531f407284e1308c3b14aab6 : TIMED_OUT in 40m 59s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cfa5243128c64bf885f67293324698ed : SUCCESS in 33m 39s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8b64ff7b9e44c7a2834127057ae5c0 : FAILURE in 27m 42s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/2988d252ab294be3b24c356fd3282e29 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":5},{"id":"f8c9e669fecfbbb59919df18d093f21cc2990022","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-13 23:44:15.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"038070f540d6212448d8ddaa53c494c87b91df86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-17 20:42:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"da469a21fd540eb73d1473bce560a34a7ba5c119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 21:20:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8766379c138f44fe9a05c99cb4e4bf53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b07fb7c4003b480d82aa2d6287163698 : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59be7c40b7c4bf7821da5baf36bd569 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab3a8dd7bc444538e8ab54574304236 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98719cb19909436587726b1355a16163 : FAILURE in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86ede6843062426daade167f61b0217e : FAILURE in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca1ad754b21b42e8b230322dcc7a709f : FAILURE in 3m 49s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2360c57abbcb439aae2fae5377087dfc : SUCCESS in 25m 47s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2803d75f376c4f51b2d6903d4fbd95f1 : FAILURE in 17m 03s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/f2b9f9f912f444bcb8ca21469e0ad04c : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":6},{"id":"b75f4f59eff05ece69cbe5e086e59eec2adb3751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-18 03:56:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3a0dd6831196e0908eab2dc6d1b1047443cdbce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 04:14:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c368f5629d427bb17afefa8ec27494\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44c63f772fb74998acbdf3f26cf95858 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba7dfb2845a429dad3f111495be7bb8 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d5a28c97794c48a6cd5e413b81d93a : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdfd3afb76f64a00ada42a5e632d419f : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b14d8185b9a4a8ba29cded2612ec0e2 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bf8d44dd6f047bea4bb00ebc1623e17 : SUCCESS in 4m 02s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/879c6ef3add743a5ab9d77649db77e84 : SUCCESS in 16m 50s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81cde06988b142a8b4e15a763ef6d490 : FAILURE in 15m 36s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/67f8c0f343c14b178df3abbac46b7b19 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":7},{"id":"325ccb79be047d22ea0014acd6fe32bce0c88728","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-18 04:25:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"996f1bdf83f1d297e73945b72abd259992214168","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-09-18 06:50:23.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"29cd371512403a96b4eed237da6a16162374a322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 06:50:53.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d19a87447d0ae0840c87ce1b5b1abab3b360529e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 07:08:57.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748589717af54fb382dacb94fa5133ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277ed262dbeb404ba5f98130ab5b8c54 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d510c050c25481b85a182b3b0cccbe3 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9f039a99a21450b832d0a4325af9b80 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6efac8b5c8424a04b3ab8db90ad667ab : SUCCESS in 3m 06s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b952c9658549413ea6a4358a8ab16653 : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":7},{"id":"a99f6d339336a177dfcc0ac9740c5ee3b097fcf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 07:08:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"ddab2ebdeeb959fac54c14d9ffe0b5a050101c6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 07:09:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835279aa17f144d18e34709f5befeffb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c18b400eebd45fc92b0f943d6fad7a3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d74365969e8c3736103b32458d53b176a3c68428","revisions":{"dfe58ff3687cbef5f7c308f2a657051f38ca5e19":{"kind":"REWORK","_number":1,"created":"2024-09-09 21:38:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/1"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:38:29.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dfe58ff3687cbef5f7c308f2a657051f38ca5e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/dfe58ff3687cbef5f7c308f2a657051f38ca5e19"}]},"branch":"refs/heads/master"},"34a7140827504f6f2a533cb8a3dd332a965c3613":{"kind":"REWORK","_number":2,"created":"2024-09-10 04:00:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/2"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-10 03:58:41.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/34a7140827504f6f2a533cb8a3dd332a965c3613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/34a7140827504f6f2a533cb8a3dd332a965c3613"}]},"branch":"refs/heads/master"},"935a09d7f6205c88d53e17e2e24809c5f194fb25":{"kind":"REWORK","_number":3,"created":"2024-09-10 15:58:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/3"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-10 15:58:35.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/935a09d7f6205c88d53e17e2e24809c5f194fb25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/935a09d7f6205c88d53e17e2e24809c5f194fb25"}]},"branch":"refs/heads/master"},"2d92a3a82ef842a6d27252b3d12712bf9cc81d5d":{"kind":"REWORK","_number":4,"created":"2024-09-13 22:48:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/4"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-13 22:47:46.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2d92a3a82ef842a6d27252b3d12712bf9cc81d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2d92a3a82ef842a6d27252b3d12712bf9cc81d5d"}]},"branch":"refs/heads/master"},"174376566103d05f8c514084c36fe998f7ab2e1c":{"kind":"REWORK","_number":5,"created":"2024-09-13 22:49:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/5"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-13 22:49:25.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/174376566103d05f8c514084c36fe998f7ab2e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/174376566103d05f8c514084c36fe998f7ab2e1c"}]},"branch":"refs/heads/master"},"a0d7c40f7265ccb45bea135f8491fe2514f9b44e":{"kind":"REWORK","_number":6,"created":"2024-09-17 20:42:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/6"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-17 20:42:43.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a0d7c40f7265ccb45bea135f8491fe2514f9b44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a0d7c40f7265ccb45bea135f8491fe2514f9b44e"}]},"branch":"refs/heads/master"},"d74365969e8c3736103b32458d53b176a3c68428":{"kind":"REWORK","_number":7,"created":"2024-09-18 03:56:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/928711/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/11/928711/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/11/928711/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/11/928711/7"}}},"commit":{"parents":[{"commit":"63c239bf3ea70ea573ed9e848660199af983217d","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63c239bf3ea70ea573ed9e848660199af983217d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-09 21:34:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-18 03:56:51.000000000","tz":-420},"subject":"Drop system scope from RBAC","message":"Drop system scope from RBAC\n\nIn new RBAC rules, all openstack services except\nironic and keystone have dropped the system scope.\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nAll the policy rules should be scoped to project and\nsystem scoped token should not be allowed to access\nthe Cyborg APIs.\n\nAs oslo.policy 4.4.0 enable the scope check and new\ndefaults, it break cyborg tests and to fix that we need\nto drop the system scope from the cyborg policies.\n\nRef:\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/6IDPXIXZ6FBKONNHFRU3ZPSXIIQJXUIE/\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/MPHSVG222OFHJL2AQD2A7CJGTH57SRCJ/\n\nChange-Id: Iafbd5470e55e10ffb6bcddd232a698eb042d7eed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d74365969e8c3736103b32458d53b176a3c68428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d74365969e8c3736103b32458d53b176a3c68428"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
