)]}'
{"id":"openstack%2Fcyborg~930919","triplet_id":"openstack%2Fcyborg~master~I7186af20055ff47ab10c6ff37bca0e3546d4affb","project":"openstack/cyborg","branch":"master","topic":"remove-eventlet-workaround","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-10-18 11:35:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7186af20055ff47ab10c6ff37bca0e3546d4affb","subject":"Remove workaround for eventlet bug #592","status":"MERGED","created":"2024-09-30 14:07:29.000000000","updated":"2024-10-18 11:36:07.000000000","submitted":"2024-10-18 11:35:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930919-remove-eventlet-workaround","meta_rev_id":"0fc813c29e2159f636a7617659cc7475cfc869ba","_number":930919,"virtual_id_number":930919,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-18 11:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":2,"date":"2024-10-17 01:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":1,"date":"2024-10-17 01:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-30 15:11:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-17 01:25:36.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2024-10-18 10:26:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"a6a14c754865b9c85f9f8a1177a2b53fd2cc8e46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-30 14:07:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12e7cb88e98efb9ae1caf4b622d070240e070581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:11:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0c2fafc51145c6873e31b6fb9f75f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aefbcfe1efa14814927357fa1683efbe : SUCCESS in 4m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/96cf5d8b81094f8f969c70a1bfa0ce30 : FAILURE in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766a260172cd4cdb81b31db79cfddb76 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b7cd6bbe9d44f298593242839c4ef0 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f7d3533bc634c2a9397a249bee330f0 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f00730e5480e457db97b55e54f77a644 : SUCCESS in 4m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0b3ca01b81c14bf9baed338a426352dc : SUCCESS in 32m 24s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cb02d8e6a2a4d5b8c2fc82fa5e857b9 : FAILURE in 21m 08s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/d4f993d6512846989c0b148c7475dbf6 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3d2b87655023b4e0fc838a65b56ed900ed4761ac","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2024-10-17 01:25:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8e78c19080c83e13a0ee39b969ad51d4d406bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 01:54:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45247ee09da749ca9c8aa3c1920e590c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7151f6d909cd4a57a7f1dd01d4e9251b : SUCCESS in 3m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da1bfcac5b114a409e48bd3c7034fbbc : FAILURE in 1m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817cd6a3d22c4890863d0ece498c25ab : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492f217214dd4e07bced8316a33840ae : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d93035cea62a496e9e8b33ee726aa918 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0da46ab338e4d9ba2c558d00f9eaa8f : SUCCESS in 3m 23s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7790e9c984f44d8a987ddd03e8809fee : SUCCESS in 16m 19s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8fbc8efe3fe4fba9d68f0f90cbbf8e8 : FAILURE in 28m 04s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/3640242e154441079f1168c789d47a05 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"97430a68da0288feaebd02c5498ab5b572e5b679","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-18 10:26:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"774fec97c9882ef11dd2f9fb398f1a8a6eeffcc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:02:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19b4748ff01f4c20b0ac55a27b525158\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6fbd4df18c8476294f53f53aff92ef2 : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4534ec68564484c8785fd55d8e161e6 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbaf1028f384806aa88cb5d994189a8 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/266e7d7c77684ff8a7a52c8b20918148 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1337df482684df097e775a39fa3716f : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/334b89090b424ee6bc74cfd5a1d89d86 : SUCCESS in 5m 59s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1a7c03d01a224cd3b60524854574aa15 : SUCCESS in 29m 38s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bccde858a8a4cad981edb205390562a : FAILURE in 28m 02s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/7b976d44db434d4b93740d6fe9d5e134 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4c5aeba52506c3653ef90a3d978e176449829610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:02:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"846ad3e8f42bf8b55342961868841efbc2d1d6d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:35:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bacfd1ee2fa242de908732fb9fb96778\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/011621f2a66e4053a9d5d6c44eb1669f : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbcd8bfc0374535a4e28a3ed7341a99 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c97686d97064ae58d251067e8ee5ec0 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cd8516c0f43406ba3b59ea80016cb68 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bdfa0512a784b999b44ea6b44070646 : SUCCESS in 3m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5192f2f12c9f4758a0431f78eedadc45 : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7972f3897bd7904b301e323a7c1545f1f6353a05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:35:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0fc813c29e2159f636a7617659cc7475cfc869ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:36:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d684d1938d4389a8cb2d324128bddf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca78e86caab4da681927d9bf6e52273 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a866dde3c57d1717540e1b642c66ca31bf1590b","revisions":{"8a866dde3c57d1717540e1b642c66ca31bf1590b":{"kind":"REWORK","_number":1,"created":"2024-09-30 14:07:29.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/19/930919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/19/930919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/19/930919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/19/930919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/19/930919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/19/930919/1"}}},"commit":{"parents":[{"commit":"d74365969e8c3736103b32458d53b176a3c68428","subject":"Drop system scope from RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d74365969e8c3736103b32458d53b176a3c68428"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-30 14:03:20.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-30 14:03:20.000000000","tz":120},"subject":"Remove workaround for eventlet bug #592","message":"Remove workaround for eventlet bug #592\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I7186af20055ff47ab10c6ff37bca0e3546d4affb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8a866dde3c57d1717540e1b642c66ca31bf1590b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8a866dde3c57d1717540e1b642c66ca31bf1590b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
