)]}'
{"id":"openstack%2Fcyborg~932254","triplet_id":"openstack%2Fcyborg~master~I78e32a520a8c7787ad3c8b79b605917648f7e6d8","project":"openstack/cyborg","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-17 01:59:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78e32a520a8c7787ad3c8b79b605917648f7e6d8","subject":"Drop deprecated onerror","status":"MERGED","created":"2024-10-12 16:19:42.000000000","updated":"2024-10-17 02:00:15.000000000","submitted":"2024-10-17 01:59:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932254","meta_rev_id":"d8c0ad0296bff0c4068a5fa752dcfe26f64b21e1","_number":932254,"virtual_id_number":932254,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-17 01:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":2,"date":"2024-10-17 01:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":1,"date":"2024-10-17 01:23:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-12 17:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-17 01:23:18.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"}],"messages":[{"id":"14cc59f87b3796e844fafed8de1a25d4f51b89d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 16:19:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8505a9fc9f890e6c8ce675d3905ac73e5227625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 17:03:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5de80e15554b8484e5144d91d3bba2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f5453c7c78c47b7879c5e7ba42bf291 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56091c2d43e489787fdbce0dc05dcbd : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d7b31054a04a459e821c66fc571104 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae7bceff367340c781346f1850df0a8c : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6813bae8cfd4479ab927ad1c65f07b25 : SUCCESS in 3m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8ce7173b57b24ce4b8d1216de29e706c : SUCCESS in 37m 03s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/967db558a07046d3a381ef1c1d1c7007 : FAILURE in 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a9e50bfeacff42bcd06d7de4a1ae1cc496e9a8e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2024-10-17 01:23:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7367a13aaa1da4d95b3ae9eb7fba9e5180db68d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 01:23:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4492e505501d6c2ebd25f38905e98a1346f0eb24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 01:59:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5bbdb11a8be4e3384d463703d32839e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39d04a5881a4ecdb293a2381531d7d1 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b3e12bab06a46c4935b44e661ac606b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2886b3ad202400eb0a7f54ccaf42240 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c00e266869f145c18b1d0f809d5ae434 : SUCCESS in 4m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/46f0c732bb5141339a48de496d60f0e2 : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1213a62db4178c06c17dfce603f8e0c78ff6f6b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 01:59:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8c0ad0296bff0c4068a5fa752dcfe26f64b21e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 02:00:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098af814c5bb49888001700313d54c3e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5e78ec6cf343d39480a1c93bf223a7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c9fc3d63ef9c897e508d1b944e31849dbf9abaa","revisions":{"2c9fc3d63ef9c897e508d1b944e31849dbf9abaa":{"kind":"REWORK","_number":1,"created":"2024-10-12 16:19:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/932254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/54/932254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/932254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/932254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/932254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/54/932254/1"}}},"commit":{"parents":[{"commit":"d74365969e8c3736103b32458d53b176a3c68428","subject":"Drop system scope from RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d74365969e8c3736103b32458d53b176a3c68428"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-12 16:18:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-12 16:18:49.000000000","tz":540},"subject":"Drop deprecated onerror","message":"Drop deprecated onerror\n\nThe onerror argument of the shutil.rmtree function was deprecated in\nPython 3.12 . Because the current value matches the actual default of\nthe function, drop the usage.\n\n[1] https://docs.python.org/3/library/shutil.html#shutil.rmtree\n\n```\nChanged in version 3.12: Added the onexc parameter, deprecated onerror.\n```\n\nChange-Id: I78e32a520a8c7787ad3c8b79b605917648f7e6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c9fc3d63ef9c897e508d1b944e31849dbf9abaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2c9fc3d63ef9c897e508d1b944e31849dbf9abaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
