)]}'
{"id":"openstack%2Fcyborg~938805","triplet_id":"openstack%2Fcyborg~master~I5c2a599679d63b32c4a0ca1754b28f10a2f62dcc","project":"openstack/cyborg","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2025-02-12 00:33:54.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-12 01:12:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c2a599679d63b32c4a0ca1754b28f10a2f62dcc","subject":"Remove six","status":"MERGED","created":"2025-01-09 14:17:11.000000000","updated":"2025-02-12 01:13:17.000000000","submitted":"2025-02-12 01:12:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938805-remove-six","meta_rev_id":"56e1574a605446f63e9b560d2d833733284a11b2","_number":938805,"virtual_id_number":938805,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 01:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2025-02-12 00:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2025-01-13 00:07:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2025-02-12 00:33:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 14:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 00:07:28.000000000","updated_by":{"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"},"reviewer":{"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"},"state":"REVIEWER"},{"updated":"2025-02-11 14:26:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"0d9c968ca04cbfc06058938775807eaa73fb1c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-09 14:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab79daa1610c12690bea9a138b4cfc652f553839","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-09 14:17:30.000000000","message":"Topic set to remove-six","accounts_in_message":[],"_revision_number":1},{"id":"bde423b11c9e4c1735ef5f1695467fb0121bd198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 14:39:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac1d4d3e56a4b0193e88ef5e8ef1559\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a5ea191f51f4b43b60fc5d6a49bb5b0 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f5355ae508420fb284ca2a2b15e7d9 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792a4307ef7c44afb13a167cc7ed85ac : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16e0176a8c594eb7861d4f1cceb87326 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/116830025349455a91b03c94b6b8e4e7 : SUCCESS in 3m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/07cc4230c038456fbc032f30607c229a : SUCCESS in 21m 34s\n- cyborg-tempest-jammy https://zuul.opendev.org/t/openstack/build/9c5f49a5338c49138a60de90d5826ea5 : SUCCESS in 18m 32s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b9b38e325d94c53b27af0c6795524fb : FAILURE in 20m 10s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/ed0cfe443b074dafbec8a221f233940f : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f29bf87f92092a632ce6426b7944bb1013226b6f","author":{"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"},"date":"2025-01-13 00:07:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"18651b0782dcc790730dfa627a4404a9aba61fec","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2025-02-12 00:33:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7f765348a8b968af0bf94ed8029ff223293c244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 00:34:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c359f607c2f2e9525f3ebdf036ae92f744beddb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:12:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a101e8b794445619e650444e86f642b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9080a866ce1b4d458cf52e175a80d194 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe85a749e884585b773a3d9c4607378 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e0ec8ab0fce4327b2fff6b1d37e255c : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa79d3120de34dd1a4352820b7304744 : SUCCESS in 2m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b62fb86b2cf84365b3c8dd5d2d354873 : SUCCESS in 32m 20s\n- cyborg-tempest-jammy https://zuul.opendev.org/t/openstack/build/b41b984354b44cf287738102ee263b61 : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"26c66bc7a1ed55f570108942e56c86162b73456f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:12:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56e1574a605446f63e9b560d2d833733284a11b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:13:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff26118e90f34a16868473c4577a3395\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c58e0849204ed6a1a296576315a7d9 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4ba33aa4b8f272281d7875d001c8e9584c14857","revisions":{"e4ba33aa4b8f272281d7875d001c8e9584c14857":{"kind":"REWORK","_number":1,"created":"2025-01-09 14:17:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/938805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/938805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/938805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/938805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/938805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/938805/1"}}},"commit":{"parents":[{"commit":"66a55c2f83f9a459a02ad781ebf89ded1f96d164","subject":"Remove default override for config options policy_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/66a55c2f83f9a459a02ad781ebf89ded1f96d164"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-09 14:15:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-09 14:15:59.000000000","tz":540},"subject":"Remove six","message":"Remove six\n\nPython 2 support was removed long time ago so we no longer need to use\nsix to maintain compatibility with Python 2.\n\nNote that six is not listed in requirements now, and this effectively\nremoves the \"hidden\" dependency.\n\nChange-Id: I5c2a599679d63b32c4a0ca1754b28f10a2f62dcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e4ba33aa4b8f272281d7875d001c8e9584c14857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e4ba33aa4b8f272281d7875d001c8e9584c14857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36714,"name":"Alexandre Detiste","display_name":"Alexandre Detiste","email":"alexandre.detiste@gmail.com","username":"adetiste"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
