)]}'
{"id":"openstack%2Fcyborg~957354","triplet_id":"openstack%2Fcyborg~master~Ie22d3eff11a8f21b401ca448e390e4ff9ec4e72a","project":"openstack/cyborg","branch":"master","topic":"release-health-check-flamingo","hashtags":[],"change_id":"Ie22d3eff11a8f21b401ca448e390e4ff9ec4e72a","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:40:17.000000000","updated":"2025-08-14 16:07:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b18ab143fbf853454871543a6b6cd3caa78b11a","_number":957354,"virtual_id_number":957354,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-14 15:51:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 15:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8cbbc14201f778ce6f4b17aaacccf8fc7aa1812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:40:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bac7177655a21b9254905640102d65204fe6a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 15:51:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa58c461aef4ffb87675ca7f46291fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3581dbd61f24837984fb3cdf5d28c3e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3586cede2e0245b38fc3ecd309f7558b : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737821a7dbc44b5ab247da95ffb04e47 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04d2d0fc273345f4aa83bf3e847a8903 : SUCCESS in 1m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb57e09626d742cda9ed291c483d6840 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba28f085b1dc4c14b6e03626c7adec8b : SUCCESS in 3m 21s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/da8decc3d6674e7ba4f812dfe9caed9f : SUCCESS in 12m 52s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/255c3a45dbde4c0292295216826481b4 : FAILURE in 12m 24s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/3524740928bb45fdbba2e7d9360499a2 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8b18ab143fbf853454871543a6b6cd3caa78b11a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 16:07:11.000000000","message":"Abandoned\n\ngate jobs are passing, patch can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6db8c38f1cd8dc692f524b15ecc49173601db34a","revisions":{"6db8c38f1cd8dc692f524b15ecc49173601db34a":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:40:17.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/957354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/54/957354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/957354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/957354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/957354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/54/957354/1"}}},"commit":{"parents":[{"commit":"0dd4b31edafda1ddc8f2054c9fe877d4d6602d39","subject":"remove jammy jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0dd4b31edafda1ddc8f2054c9fe877d4d6602d39"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:40:07.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:40:15.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Ie22d3eff11a8f21b401ca448e390e4ff9ec4e72a\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6db8c38f1cd8dc692f524b15ecc49173601db34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6db8c38f1cd8dc692f524b15ecc49173601db34a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
