)]}'
{"id":"openstack%2Fcyborg~966854","triplet_id":"openstack%2Fcyborg~master~I25595fe1beeeb216053464ec1d36887896557fa3","project":"openstack/cyborg","branch":"master","topic":"requirements","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-02-04 15:08:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I25595fe1beeeb216053464ec1d36887896557fa3","subject":"Use the correct package name for microversion-parse","status":"MERGED","created":"2025-11-12 14:11:45.000000000","updated":"2026-02-10 10:46:12.000000000","submitted":"2026-02-09 18:45:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"966854-requirements","meta_rev_id":"847a3fc3eba3038d26957bce77501a7798311bbf","_number":966854,"virtual_id_number":966854,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 18:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-02-04 15:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-01-27 17:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":1,"date":"2026-02-02 05:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-02-09 18:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 16:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-17 14:27:14.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2026-02-02 05:36:08.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"bf872e4ccc973cde8dc57a497f9a5634f5104fcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-12 14:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f8e32ed2b8e74a77ff82df9d6a66007d03589e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 16:07:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c8d31b9e8394443800984d3647b6bff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02b56f2a7fb145d7a0063df1d4b88661 : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/efe0dacea6b64840b36ef34c251d5fc0 : POST_FAILURE in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d4b41893a3240e0adec8f4ec410e406 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d39749dc964803b25fdeeb9c65f84a : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b115a64409b942549f0dd36844ae5bcf : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5da1c8d89dba4daeb99b4f7ec5cf3b9b : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5df075951fc6d4b2c1f3cf98dd3b98fcfc4ab582","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-12 16:41:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0b35296fc7abd9c0c8e39f2ba14b6b4c417dbf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:23:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d06d3d42ccd4fc9b658ac471df216bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3665687815574d34b0f530d6d2027b5a : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f243c7c8500149308f6a37447771122c : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31fe571449a4b91a04c25bc8a3c3ffa : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99046a107ad49988f5962684b4f8793 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4dbec4221424a2db5506ce5030eab5d : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/208365cd9a484df0a5bdc21c484d518e : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b42ad6f226b198aa4aa1f1fc1782a36268d7194d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-08 17:23:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"21d70420f31835f9d91ac0295acdb6bdaff4849c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 17:37:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1eedbecfc5469d96df8b6f26a56f44\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34f53d246ddb441dabb937180655a433 : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/def2da67aaa443a0915e38198e18cd31 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ae09a02a7c459f94bc83af312ef5f5 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d5b4f4731d4311889cb7b25eb2b37c : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0b8fc80af8c43fb890657a7c99777f5 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0025dc2d0abf4977aae8840165046ca0 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":2},{"id":"91136f6f7d32f303174658e4528bc0c98ff84603","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-12-17 14:27:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5bbea9c3f54cab8038f9f8600a06503d39a85908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-27 17:59:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1816de64058ed90eb0f56db3a0185bd8598f7da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:12:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/355a97a2b28749fc81b4fcfc4e9286cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97eee9ed515848b3abe6d4896d73e397 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/836a597c31f3417a8cccca1f775a73ec : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69387cb3d2c24049981f147b059e58c6 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ccfb066f5240ec9d18599bd8c59eaa : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6291ca66d4664d32a1ada2d2a9da63ac : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86f7983bd308426da667858782dff950 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c7f325f02abc1aec2cabea219b2c9546d9351a9e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-02 05:36:08.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9ef225e956c79bf59aaea9bf387c1e61887df6d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-04 15:08:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a02a5dc770b3ab6ea7b84a232154fe3be6afd99b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-09 18:32:00.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ef6e0a407fd5487e8ea4bf925e867b2f9ed04126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 18:32:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d202ef33bd0b4faae595ae00bc8cc8b8287a485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 18:45:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9509dcc2fb654a8eb09270bfe4d498d7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b47eba8571d342d3a7a3b52270d0fb04 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59a7b8a095846d2b6417d7bc1fd2729 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09eee78c2cf4149b953e421fd80d424 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/664aef94974646baba4ad8d3302c8829 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f8306d56fa4462089639325aca4f433 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":3},{"id":"ada85d9d2e8a2f62fb37252c362e5d0eac34b894","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 18:45:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9041ee1938c1958ee19a52953d9f1c8c69fa4788","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 18:45:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf15940ce0204d49bed96343c7203fdc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a093f92951e4a2191490fed04384fc1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3},{"id":"847a3fc3eba3038d26957bce77501a7798311bbf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-10 10:46:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22d6ded9fd97bdd6c49c619fd9cdd6b650a0b8d0","revisions":{"aa55d28364ed3e35ab5623389b34ddd1569b84d4":{"kind":"REWORK","_number":1,"created":"2025-11-12 14:11:45.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/54/966854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/54/966854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/54/966854/1"}}},"commit":{"parents":[{"commit":"e408b4c1b7e233a954da1ab5b6119014f8d95b26","subject":"remove max version desc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e408b4c1b7e233a954da1ab5b6119014f8d95b26"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-12 14:11:24.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-12 14:11:24.000000000","tz":0},"subject":"use the correct package name for microversion-parse","message":"use the correct package name for microversion-parse\n\nAs of Ib97eea1037fac541c8ebadc526ae9c3bb160f685 the\nrequirements check job nolonger uses pkg_resources\nto normalise the package names, this highlighted that\nwe are not using the actuall name and were relying on\nthat normalization for microversion-parse.\n\nThis change fixes that by using the correct name.\n\nChange-Id: I25595fe1beeeb216053464ec1d36887896557fa3\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/aa55d28364ed3e35ab5623389b34ddd1569b84d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/aa55d28364ed3e35ab5623389b34ddd1569b84d4"}]},"branch":"refs/heads/master"},"620e66f9e299dc185724372ec13f501f5945cc2f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-12-08 17:23:20.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/54/966854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/54/966854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/54/966854/2"}}},"commit":{"parents":[{"commit":"75467de9f767fe552c891a262891f62b38f63ab6","subject":"Refactor session handling and align test contexts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/75467de9f767fe552c891a262891f62b38f63ab6"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-12 14:11:24.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-12-08 17:22:25.000000000","tz":0},"subject":"Use the correct package name for microversion-parse","message":"Use the correct package name for microversion-parse\n\nAs of Ib97eea1037fac541c8ebadc526ae9c3bb160f685 the\nrequirements check job nolonger uses pkg_resources\nto normalise the package names, this highlighted that\nwe are not using the actuall name and were relying on\nthat normalization for microversion-parse.\n\nThis change fixes that by using the correct name.\n\nChange-Id: I25595fe1beeeb216053464ec1d36887896557fa3\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/620e66f9e299dc185724372ec13f501f5945cc2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/620e66f9e299dc185724372ec13f501f5945cc2f"}]},"branch":"refs/heads/master"},"22d6ded9fd97bdd6c49c619fd9cdd6b650a0b8d0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-27 17:59:43.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/54/966854/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/54/966854/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/54/966854/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/54/966854/3"}}},"commit":{"parents":[{"commit":"ef5fc870846075e578582597fd010e1f1122b6f3","subject":"Remove deprecated get_legacy_facade usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ef5fc870846075e578582597fd010e1f1122b6f3"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-12 14:11:24.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-01-27 17:58:53.000000000","tz":0},"subject":"Use the correct package name for microversion-parse","message":"Use the correct package name for microversion-parse\n\nAs of Ib97eea1037fac541c8ebadc526ae9c3bb160f685 the\nrequirements check job nolonger uses pkg_resources\nto normalise the package names, this highlighted that\nwe are not using the actuall name and were relying on\nthat normalization for microversion-parse.\n\nThis change fixes that by using the correct name.\n\nChange-Id: I25595fe1beeeb216053464ec1d36887896557fa3\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/22d6ded9fd97bdd6c49c619fd9cdd6b650a0b8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/22d6ded9fd97bdd6c49c619fd9cdd6b650a0b8d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
